)]}'
{"id":"openstack%2Freleases~695979","triplet_id":"openstack%2Freleases~master~Ic0b41a4c23ea293ae2bbbd9ae7d3d3adc180eb34","project":"openstack/releases","branch":"master","topic":"oslo.messaging-nats-driver","hashtags":[],"change_id":"Ic0b41a4c23ea293ae2bbbd9ae7d3d3adc180eb34","subject":"Create feature branch for nats-driver work in oslo.messaging","status":"MERGED","created":"2019-11-25 20:09:02.000000000","updated":"2019-11-26 10:41:03.000000000","submitted":"2019-11-26 10:41:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695979-1574764863788-ded4c5e8","meta_rev_id":"66cf35c121871d2850a628d0bec140cb333d18d5","_number":695979,"virtual_id_number":695979,"owner":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-26 10:41:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-11-25 20:13:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-11-25 20:17:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-11-26 10:29:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-11-26 10:29:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 20:13:36.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-11-25 20:17:55.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-11-26 10:29:42.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-11-26 10:41:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7f0bef2ff78d6873e7649a8f0776b0f896a214a5","author":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"date":"2019-11-25 20:09:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b92c7c531a4a0bfcf03fae331fad7c5b49ebcc6","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-11-25 20:13:36.000000000","message":"Patch Set 1: Code-Review+1\n\nThis is how we agreed to proceed with this newly proposed driver in the Oslo virtual PTG today.","accounts_in_message":[],"_revision_number":1},{"id":"8da7de62330b3aee08c3a63c899a0723daaca06e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-25 20:17:55.000000000","message":"Patch Set 1: Code-Review+2\n\nI don\u0027t know the scope of the work involved, but just want to mention that other teams have found feature branches to be more work than they had anticipated. They can work well and be useful to segregate some major, long term changes, so there is certainly some value. Just a note of caution. ;)","accounts_in_message":[],"_revision_number":1},{"id":"f57cb5b39152c7befc5e80c61b62cc46c20433e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 20:19:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2952f02120f441beb10e8607b1114df9 : SUCCESS in 9m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/edf3e0c587fb43489edb6b0b6f0f3bf0 : SUCCESS in 4m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cbe4435da9a148f38998c695e33f7fc7 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"617327ac97f94acd3ed0cbc36220634456dae4c9","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-11-25 20:35:14.000000000","message":"Patch Set 1:\n\n\u003e I don\u0027t know the scope of the work involved, but just want to\n \u003e mention that other teams have found feature branches to be more\n \u003e work than they had anticipated. They can work well and be useful to\n \u003e segregate some major, long term changes, so there is certainly some\n \u003e value. Just a note of caution. ;)\n\nThere\u0027s kind of two reasons we went with a feature branch: 1) We\u0027re not clear about the viability of the new driver and don\u0027t want to merge things into master that we will then be on the hook to support. 2) We think this is probably lower overhead than developing the driver in a completely separate repo. Also, the separate repo doesn\u0027t allow changes to the oslo.messaging driver interface, if any are needed.\n\nHopefully it won\u0027t be too difficult to maintain the branch. There isn\u0027t _that_ much major work going on in oslo.messaging these days so it shouldn\u0027t be a merge conflict nightmare. As Ken said in the meeting, we\u0027ll jump off that bridge if/when we come to it. :-)","accounts_in_message":[],"_revision_number":1},{"id":"448dd083aa7cd6920f4a88914239668269f894d2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-26 10:29:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"495a75f84fc488de7b56740dd57d7ee1d5c11605","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 10:29:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f3232f32fe9108ebe5920021859f783e04787df7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 10:41:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8087adad5d4defbf9b6069de5193ef : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be952fca182a43268e6607dafab9b0bd : SUCCESS in 8m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/84e48f224867413ebec8de581f7c5093 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":1},{"id":"581af4443fcceb04be56fd47b482c27e09062327","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 10:41:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6f23a48c7a163ae4494e146e7224aead026d02ce","revisions":{"6f23a48c7a163ae4494e146e7224aead026d02ce":{"kind":"REWORK","_number":1,"created":"2019-11-25 20:09:02.000000000","uploader":{"_account_id":8770,"name":"Ken Giusti","email":"kgiusti@gmail.com","username":"kgiusti"},"ref":"refs/changes/79/695979/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/79/695979/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/695979/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/695979/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/79/695979/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/79/695979/1"}}},"commit":{"parents":[{"commit":"8568a30288fbe619de74b7989d24efac9485daa5","subject":"Merge \"Release OpenStack-Ansible Rocky\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8568a30288fbe619de74b7989d24efac9485daa5"}]}],"author":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2019-11-25 16:41:07.000000000","tz":-300},"committer":{"name":"Kenneth Giusti","email":"kgiusti@gmail.com","date":"2019-11-25 20:06:46.000000000","tz":-300},"subject":"Create feature branch for nats-driver work in oslo.messaging","message":"Create feature branch for nats-driver work in oslo.messaging\n\nCreates a temporary branch dedicated to the development of a\nNATS based driver.\n\nSee [0] for the launchpad entry and [1] for the oslo-spec proposal.\n\n[0] https://blueprints.launchpad.net/oslo.messaging/+spec/nats-transport\n[1] https://review.opendev.org/#/c/692784/\n\nChange-Id: Ic0b41a4c23ea293ae2bbbd9ae7d3d3adc180eb34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f23a48c7a163ae4494e146e7224aead026d02ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6f23a48c7a163ae4494e146e7224aead026d02ce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
