)]}'
{"id":"openstack%2Freleases~696095","triplet_id":"openstack%2Freleases~master~Iaf8ffcb691f14e83843bf2079fabaa87776f0526","project":"openstack/releases","branch":"master","topic":"check-approval","hashtags":[],"change_id":"Iaf8ffcb691f14e83843bf2079fabaa87776f0526","subject":"Enable check-approval in experimental pipeline","status":"MERGED","created":"2019-11-26 14:28:25.000000000","updated":"2019-12-13 15:16:11.000000000","submitted":"2019-12-13 15:16:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696095-1576250171865-71032318","meta_rev_id":"5679b4baa7ba444b5f4b74e6bc67825bff18c7e8","_number":696095,"virtual_id_number":696095,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-13 15:16:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-05 16:54:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":2,"date":"2019-12-05 16:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":1,"date":"2019-12-05 16:36:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-05 16:36:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-12-05 16:54:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 17:55:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-11-27 11:11:55.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-12-03 19:46:27.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-12-05 16:54:23.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-13 15:16:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"55a21cba5475063c4f862121001fe52138bd3b3c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-26 14:28:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"91ff958d49028c578c3a71f349919ee087913b59","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 14:28:41.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/releases on branch master.  The error was:\n\n  unhashable type: \u0027list\u0027","accounts_in_message":[],"_revision_number":1},{"id":"a5fb7733dba3faa01ef0d9a74964860b1c6a4a07","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-26 14:57:30.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54feb81854d72a72397f7e26e3697141ad611a9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:15:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9472e7ae4d604b3398b33e721da24051 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b772de899cbd4d5297de1e5f2a114610 : SUCCESS in 4m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3837b281b564455fb5335252f4b094aa : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edbdcc3f41ea46b0923aea78c9891a8d : SUCCESS in 8m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/be88a0bdb03041288820101a4084f919 : SUCCESS in 5m 01s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/335676c65c7d4003a864276699d9ef03 : SUCCESS in 7m 12s","accounts_in_message":[],"_revision_number":2},{"id":"20dda76e97f5471b64e88916d1aff1b1726dfdb9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-26 15:54:07.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"60bc4a237421e8805aafcdb8d48f56a395e7044c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-26 15:55:30.000000000","message":"Patch Set 3:\n\nLooks like classic tox jobs won\u0027t run on the executor as-is, so let\u0027s try out executing on a worker node for now.","accounts_in_message":[],"_revision_number":3},{"id":"346746964cbc4620c09c6bfc583900bc88009050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:08:33.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c973b26eec4e45b6adb1e61cbd6cfcf8 : SUCCESS in 4m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6c5e77ad3b740e68ca8882b1e2e58e0 : SUCCESS in 4m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/736054718c3749a1b5e5be90edcea91f : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0f0f95d4b9c45698bbf45fd980e9ff9 : SUCCESS in 9m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/465557bb792a4fecb2774727847f979e : SUCCESS in 4m 55s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7df2c75492e14ca4b268593cfca2fbe7 : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":3},{"id":"7dee4f43c69e48640afa878ab1932db1113a4cd8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-26 17:55:47.000000000","message":"Patch Set 3: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"a208ac5ca4e26eb3a99e4d63868a9156e6d360e0","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-11-27 11:11:55.000000000","message":"Patch Set 3: Code-Review+2\n\nThe check-approval is well triggered in the experimental pipeline (see https://review.opendev.org/#/c/696104/2). For me it\u0027s good enough.","accounts_in_message":[],"_revision_number":3},{"id":"74b997fcd344deb72db79780f70f38f44f260a13","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-03 19:46:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"041fa04afab09f0d9df4871b8cec4dd30f8a73a9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-05 16:36:07.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"9f8de12a2aac245863458a23de6d78936416b106","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 16:52:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5f88a951f274f3589b0e455d25b60cd : SUCCESS in 4m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cba2232e62e34624877efafa4ba3c85d : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/211bfd5d253f464490799e32a2e92930 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/815f4f9e9fbb4ce7a23660b02975b89b : SUCCESS in 10m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/10a62f413381498b85d49f4a1c4a31bb : SUCCESS in 7m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ef0c496c0d9e473290c4d912e3e44981 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":4},{"id":"e7a6f7c8022fe126671372791b584a0bafd0c04d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-05 16:54:23.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"9da6033c5f2cc0878d1b71a9b702005125587dfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 15:04:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"02679858ffd29cd487e4c898acfc45c7e7f6a48e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 15:16:11.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b4dcc260d5d42aead40dcdd1c974772 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb95f206544b4691a2df7f051ed32319 : SUCCESS in 6m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0bc50aec7f7d4fea8bd5f10fb6166e6e : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cecc419783a9429984c60806018fc56e : SUCCESS in 9m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d59b9f878e424f2e9f72df5c811cfbd1 : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":4},{"id":"d6fdd43c17a952e43661f0fbeee9cc072aee8250","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 15:16:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"66f14e1d32e39fe0fa599deafed529467070d541","revisions":{"cc49b0903462dcd73a8e8a1b65e89536c6da466c":{"kind":"REWORK","_number":1,"created":"2019-11-26 14:28:25.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/95/696095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/696095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/696095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/696095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/696095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/696095/1"}}},"commit":{"parents":[{"commit":"71bb3db5d4a18a71d0bf121d56413d1f3988b120","subject":"Introduce tool to check PTL/liaison approval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71bb3db5d4a18a71d0bf121d56413d1f3988b120"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 14:25:31.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 14:27:16.000000000","tz":60},"subject":"Enable check-approval in experimental pipeline","message":"Enable check-approval in experimental pipeline\n\nDefine a check-approval job that runs check_approval.py, and\nenable it in the experimental pipeline for now.\n\nThe job will run on the local executor, with an aggressive timeout\n(it should only take a couple of seconds anyway).\n\nChange-Id: Iaf8ffcb691f14e83843bf2079fabaa87776f0526\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cc49b0903462dcd73a8e8a1b65e89536c6da466c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cc49b0903462dcd73a8e8a1b65e89536c6da466c"}]},"branch":"refs/heads/master"},"abe4f7adfa639614207badc10f67a1414ab46f25":{"kind":"REWORK","_number":2,"created":"2019-11-26 14:57:30.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/95/696095/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/696095/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/696095/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/696095/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/696095/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/696095/2"}}},"commit":{"parents":[{"commit":"71bb3db5d4a18a71d0bf121d56413d1f3988b120","subject":"Introduce tool to check PTL/liaison approval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/71bb3db5d4a18a71d0bf121d56413d1f3988b120"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 14:25:31.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 14:57:12.000000000","tz":60},"subject":"Enable check-approval in experimental pipeline","message":"Enable check-approval in experimental pipeline\n\nDefine a check-approval job that runs check_approval.py, and\nenable it in the experimental pipeline for now.\n\nThe job will run on the local executor, with an aggressive timeout\n(it should only take a couple of seconds anyway).\n\nChange-Id: Iaf8ffcb691f14e83843bf2079fabaa87776f0526\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/abe4f7adfa639614207badc10f67a1414ab46f25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/abe4f7adfa639614207badc10f67a1414ab46f25"}]},"branch":"refs/heads/master"},"95841b6cdec9732fa3d05d23baa75bd35f6a7e12":{"kind":"REWORK","_number":3,"created":"2019-11-26 15:54:07.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/95/696095/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/696095/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/696095/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/696095/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/696095/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/696095/3"}}},"commit":{"parents":[{"commit":"2f38d394c9afc8d76d56f1662f0d0eccbc2b8d74","subject":"Introduce tool to check PTL/liaison approval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2f38d394c9afc8d76d56f1662f0d0eccbc2b8d74"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 14:25:31.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 15:52:52.000000000","tz":60},"subject":"Enable check-approval in experimental pipeline","message":"Enable check-approval in experimental pipeline\n\nDefine a check-approval job that runs check_approval.py, and\nenable it in the experimental pipeline for now.\n\nAs a first step the job will run as a classic tox job on a worker\nnode, with an aggressive 6-min timeout.\n\nChange-Id: Iaf8ffcb691f14e83843bf2079fabaa87776f0526\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95841b6cdec9732fa3d05d23baa75bd35f6a7e12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95841b6cdec9732fa3d05d23baa75bd35f6a7e12"}]},"branch":"refs/heads/master"},"66f14e1d32e39fe0fa599deafed529467070d541":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-05 16:36:07.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/95/696095/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/696095/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/696095/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/696095/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/696095/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/696095/4"}}},"commit":{"parents":[{"commit":"09e8ffb3b12c5bb5ae76c6e39eb826ab2189b3c5","subject":"Introduce tool to check PTL/liaison approval","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/09e8ffb3b12c5bb5ae76c6e39eb826ab2189b3c5"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 14:25:31.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-05 16:35:15.000000000","tz":60},"subject":"Enable check-approval in experimental pipeline","message":"Enable check-approval in experimental pipeline\n\nDefine a check-approval job that runs check_approval.py, and\nenable it in the experimental pipeline for now.\n\nAs a first step the job will run as a classic tox job on a worker\nnode, with an aggressive 6-min timeout.\n\nChange-Id: Iaf8ffcb691f14e83843bf2079fabaa87776f0526\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66f14e1d32e39fe0fa599deafed529467070d541"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66f14e1d32e39fe0fa599deafed529467070d541"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
