)]}'
{"id":"openstack%2Freleases~696104","triplet_id":"openstack%2Freleases~master~Ib1c7ae4f7af8dd109ff815db9df4fb6d76f85a3e","project":"openstack/releases","branch":"master","topic":"check-approval","hashtags":[],"change_id":"Ib1c7ae4f7af8dd109ff815db9df4fb6d76f85a3e","subject":"[DNM] Test check_approval job on fake release","status":"ABANDONED","created":"2019-11-26 15:10:35.000000000","updated":"2020-01-27 11:13:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"87605f16c527ce8e9bfb5ec99a85d0c094972966","_number":696104,"virtual_id_number":696104,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-01-27 11:11:50.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-01-27 10:54:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-27 11:12:34.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-01-27 10:54:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"450a1b5c5ccea8d6052d233071f7f24f0095bb47","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-26 15:10:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f80644086060bb20c5816a2853c5d28b81d7e29c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-26 15:17:54.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"37130566eb253cf803be5b82ec50c8d6df5115ee","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:21:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (experimental pipeline).\n\n- releases-check-approval https://zuul.opendev.org/t/openstack/build/50b8660fac5b4d98a77f3c11392c5b9f : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1},{"id":"632dd6ea94e9ecf489a9c110bab78b6e7a4d4666","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 15:24:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d8989b47d1347d5b064570747d278a9 : SUCCESS in 8m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f9f69a14ca694530847296778869f247 : SUCCESS in 4m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1cf36196dbc04f418d0858a305893539 : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8ccc727f71c9a27e0ae0e4c2697d2e58070a3e7d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-26 15:54:07.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"95837b3f53e8b3d64acbdf3d0f9a1f8655674f01","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-11-26 15:55:47.000000000","message":"Patch Set 2:\n\ncheck experimental","accounts_in_message":[],"_revision_number":2},{"id":"933e11cee5344b91891fc4c7e798eee12d836fdb","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:04:53.000000000","message":"Patch Set 2:\n\nBuild failed (experimental pipeline).\n\n- releases-check-approval https://zuul.opendev.org/t/openstack/build/5eaead706b4e4bad90f70c6df43696f4 : FAILURE in 6m 43s","accounts_in_message":[],"_revision_number":2},{"id":"1dda2bd0956e527c6dbb44f44c519745abd31312","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 16:12:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/821d7c44240344ddac3535af67c2cf8e : SUCCESS in 8m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a407d7be6914445891e78dbc3a290368 : SUCCESS in 5m 43s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/faa4a1a1cd7f4fa1836aa3f9561d2cdb : SUCCESS in 4m 52s","accounts_in_message":[],"_revision_number":2},{"id":"da0884690bdb541c633bc5e537dc1a5eb54aa9cc","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-11-27 11:12:34.000000000","message":"Patch Set 2:\n\nWe should probably do something about that job failure, in case we want to trigger the experimental job on release-test in the future! :)","accounts_in_message":[],"_revision_number":2},{"id":"54648626f6d7b81808ce2a086cb30290ec62b48e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-05 16:36:07.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"fecff9c7a5ebff69814866ec41739fd64a1d5a48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 16:50:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01449c9d4e5940ffaac5017c00c52312 : SUCCESS in 9m 56s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d13f7a046a15470ea6edc2c22dc91ece : SUCCESS in 5m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8fdaffe4bed04330b08b567db6dc19b9 : SUCCESS in 6m 17s","accounts_in_message":[],"_revision_number":3},{"id":"c31761794ae577ac9d6dd834938fa91b8c99069b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-06 10:19:17.000000000","message":"Patch Set 3:\n\ncheck experimental","accounts_in_message":[],"_revision_number":3},{"id":"ed4e78edc8a8c92da9fe8159629f2238a9fcf6f1","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 10:24:04.000000000","message":"Patch Set 3:\n\nBuild failed (experimental pipeline).\n\n- releases-check-approval https://zuul.opendev.org/t/openstack/build/e51b35307eb844dfa7d2050725a9c1ec : FAILURE in 4m 17s","accounts_in_message":[],"_revision_number":3},{"id":"d0c210bbc7d6f0996f8ac154b352286d33b5230e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-13 15:20:49.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"3bd461ca8e68e6f662df1f9814fcd2433c1ecebc","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-13 15:21:14.000000000","message":"Patch Set 4:\n\ncheck experimental","accounts_in_message":[],"_revision_number":4},{"id":"0991aaa1c816e1693f1977996d1a9c2d8d19c0b1","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 15:22:54.000000000","message":"Patch Set 4:\n\nBuild succeeded (experimental pipeline).\n\n- releases-check-approval https://zuul.opendev.org/t/openstack/build/4056ca3ee8b247ebbe1cbb1474191c16 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":4},{"id":"29e78e452999a74b7b828746de877467660a801e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 15:35:40.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f680f7a8a44176ab159d8792502f4b : SUCCESS in 9m 22s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dab32a5e023342fd8ff8544dad687dba : SUCCESS in 6m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cd2bd857cb2346bb89dab748f08f1670 : SUCCESS in 7m 46s","accounts_in_message":[],"_revision_number":4},{"id":"ce8368b8346869aecf804c2c68104b0131c1388b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-27 10:34:59.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"9545e1c99b968f53b2bc815e1fa103f0dc098840","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-27 10:36:34.000000000","message":"Patch Set 5:\n\ncheck experimental","accounts_in_message":[],"_revision_number":5},{"id":"b6155f9e0efe535c60019a725a916a162b5c0d34","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 10:37:38.000000000","message":"Patch Set 5:\n\nBuild failed (experimental pipeline).\n\n- check-release-approval https://zuul.opendev.org/t/openstack/build/cb9bf31f97e84893895ec50a80db23f1 : FAILURE in 41s","accounts_in_message":[],"_revision_number":5},{"id":"7870a0652d97ae3105c04ecbd07b03b3d81b5b93","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-27 10:43:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"fd7bc02f4479ddc7a59f808dd6319793e4754c50","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-27 10:44:26.000000000","message":"Patch Set 6:\n\ncheck experimental","accounts_in_message":[],"_revision_number":6},{"id":"a63ee9baac8175f63bab438afae32653c5055a18","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 10:45:32.000000000","message":"Patch Set 6:\n\nBuild failed (experimental pipeline).\n\n- check-release-approval https://zuul.opendev.org/t/openstack/build/53e2e6d1f724410a80fd538793c54ab5 : FAILURE in 42s","accounts_in_message":[],"_revision_number":6},{"id":"7503e89eccb7b42cc21ac9599388ba9ebe9c2999","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 10:54:38.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/beee6becd3864dc485459da9ffb1031c : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c749e3b07c7e4269a75be1c3a1ff359c : SUCCESS in 10m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/92f7a1cb69e842568cde3450b0861781 : SUCCESS in 5m 47s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/060f0e215b0a482fb253452064d4ee42 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":6},{"id":"d92853af89ceeecf6a411d45f4a6827ebc8fd6e3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-27 11:11:50.000000000","message":"Patch Set 6:\n\nFix for corner issue included in https://review.opendev.org/704306","accounts_in_message":[],"_revision_number":6},{"id":"2ea9e9b877d8f5afdb1d45ea00087c25b5373109","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-27 11:13:30.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7147c24e963e4add4fe440947b3f8a2a55559fee","revisions":{"d7fbf20943e4a4d86fbbc4e2882551ccca276aa8":{"kind":"REWORK","_number":1,"created":"2019-11-26 15:10:35.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/04/696104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/04/696104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/04/696104/1"}}},"commit":{"parents":[{"commit":"abe4f7adfa639614207badc10f67a1414ab46f25","subject":"Enable check-approval in experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/abe4f7adfa639614207badc10f67a1414ab46f25"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 15:09:09.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 15:09:09.000000000","tz":60},"subject":"[DNM] Test check_approval job on fake release","message":"[DNM] Test check_approval job on fake release\n\nFake ussuri-1 release to check the new check-approval experimental job.\n\nChange-Id: Ib1c7ae4f7af8dd109ff815db9df4fb6d76f85a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7fbf20943e4a4d86fbbc4e2882551ccca276aa8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d7fbf20943e4a4d86fbbc4e2882551ccca276aa8"}]},"branch":"refs/heads/master"},"091ccb6d7cecc2429bdf9029528ce64d9cd180d1":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-11-26 15:54:07.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/04/696104/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/04/696104/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/04/696104/2"}}},"commit":{"parents":[{"commit":"95841b6cdec9732fa3d05d23baa75bd35f6a7e12","subject":"Enable check-approval in experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/95841b6cdec9732fa3d05d23baa75bd35f6a7e12"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 15:09:09.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 15:53:56.000000000","tz":60},"subject":"[DNM] Test check_approval job on fake release","message":"[DNM] Test check_approval job on fake release\n\nFake ussuri-1 release to check the new check-approval experimental job.\n\nChange-Id: Ib1c7ae4f7af8dd109ff815db9df4fb6d76f85a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/091ccb6d7cecc2429bdf9029528ce64d9cd180d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/091ccb6d7cecc2429bdf9029528ce64d9cd180d1"}]},"branch":"refs/heads/master"},"dd6dd6c11fc5bdc7c1ad309801ca199aec438c71":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-12-05 16:36:07.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/04/696104/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/04/696104/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/04/696104/3"}}},"commit":{"parents":[{"commit":"66f14e1d32e39fe0fa599deafed529467070d541","subject":"Enable check-approval in experimental pipeline","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/66f14e1d32e39fe0fa599deafed529467070d541"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 15:09:09.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-05 16:35:15.000000000","tz":60},"subject":"[DNM] Test check_approval job on fake release","message":"[DNM] Test check_approval job on fake release\n\nFake ussuri-1 release to check the new check-approval experimental job.\n\nChange-Id: Ib1c7ae4f7af8dd109ff815db9df4fb6d76f85a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dd6dd6c11fc5bdc7c1ad309801ca199aec438c71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dd6dd6c11fc5bdc7c1ad309801ca199aec438c71"}]},"branch":"refs/heads/master"},"dd7514667f45b646a9bb66de6f3d866d4d680461":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-13 15:20:49.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/04/696104/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/04/696104/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/04/696104/4"}}},"commit":{"parents":[{"commit":"f9a0d140fc436d66ae321fd54ca4c36ba6409bb4","subject":"Execute check_approval directly on Zuul node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9a0d140fc436d66ae321fd54ca4c36ba6409bb4"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 15:09:09.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-13 15:20:49.000000000","tz":0},"subject":"[DNM] Test check_approval job on fake release","message":"[DNM] Test check_approval job on fake release\n\nFake ussuri-1 release to check the new check-approval experimental job.\n\nChange-Id: Ib1c7ae4f7af8dd109ff815db9df4fb6d76f85a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dd7514667f45b646a9bb66de6f3d866d4d680461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dd7514667f45b646a9bb66de6f3d866d4d680461"}]},"branch":"refs/heads/master"},"5e14d4b4fe9b4ab654df061665038be24f26a9b9":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-01-27 10:34:59.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/04/696104/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/04/696104/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/04/696104/5"}}},"commit":{"parents":[{"commit":"a56bc4f46cd0bb919b5578ee97dc0fa134d5b893","subject":"Enable check-release-approval on experimental","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a56bc4f46cd0bb919b5578ee97dc0fa134d5b893"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 15:09:09.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-27 10:34:59.000000000","tz":0},"subject":"[DNM] Test check_approval job on fake release","message":"[DNM] Test check_approval job on fake release\n\nFake ussuri-1 release to check the new check-approval experimental job.\n\nChange-Id: Ib1c7ae4f7af8dd109ff815db9df4fb6d76f85a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5e14d4b4fe9b4ab654df061665038be24f26a9b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5e14d4b4fe9b4ab654df061665038be24f26a9b9"}]},"branch":"refs/heads/master"},"7147c24e963e4add4fe440947b3f8a2a55559fee":{"kind":"REWORK","_number":6,"created":"2020-01-27 10:43:49.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/04/696104/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/04/696104/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/04/696104/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/04/696104/6"}}},"commit":{"parents":[{"commit":"a56bc4f46cd0bb919b5578ee97dc0fa134d5b893","subject":"Enable check-release-approval on experimental","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a56bc4f46cd0bb919b5578ee97dc0fa134d5b893"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-11-26 15:09:09.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-27 10:43:24.000000000","tz":60},"subject":"[DNM] Test check_approval job on fake release","message":"[DNM] Test check_approval job on fake release\n\nFake ussuri-1 release to check the new check-approval experimental job.\n\nChange-Id: Ib1c7ae4f7af8dd109ff815db9df4fb6d76f85a3e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7147c24e963e4add4fe440947b3f8a2a55559fee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7147c24e963e4add4fe440947b3f8a2a55559fee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
