)]}'
{"id":"openstack%2Freleases~696155","triplet_id":"openstack%2Freleases~master~I3a9b2ccb88c92a2da92bc1bbdae8380a5a835d71","project":"openstack/releases","branch":"master","topic":"tripleo/stein","hashtags":[],"change_id":"I3a9b2ccb88c92a2da92bc1bbdae8380a5a835d71","subject":"update release tags for tripleo stein","status":"MERGED","created":"2019-11-26 18:20:55.000000000","updated":"2019-12-12 10:06:01.000000000","submitted":"2019-12-12 10:06:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696155-1576145161838-0c0e24f5","meta_rev_id":"589e2bf3f1aa2c382d07d976813cb70d019aa423","_number":696155,"virtual_id_number":696155,"owner":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 10:06:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2019-12-11 09:27:02.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-03 14:17:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":2,"date":"2019-12-12 09:46:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-12-03 02:34:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2019-12-02 23:55:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":1,"date":"2019-12-03 00:06:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":1,"date":"2019-12-11 10:07:50.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":1,"date":"2019-12-12 09:46:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"value":0,"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"value":0,"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 18:21:09.000000000","updated_by":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"reviewer":{"_account_id":3153,"name":"Emilien Macchi","email":"emilien@redhat.com","username":"emilienm"},"state":"REVIEWER"},{"updated":"2019-11-26 18:58:53.000000000","updated_by":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"reviewer":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"state":"REVIEWER"},{"updated":"2019-12-03 00:06:09.000000000","updated_by":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"reviewer":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"state":"REVIEWER"},{"updated":"2019-12-03 02:34:34.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2019-12-03 14:17:17.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-11 09:27:02.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-12-11 10:07:50.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-12-12 09:46:42.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-12-12 10:06:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"57afa13b3b73394fd9122b86f62b9751276c6fcc","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-11-26 18:20:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f79d081bcd18a87eadfc3e85ac7fd5e2daf4f7b7","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-11-26 18:21:31.000000000","message":"Topic set to tripleo/stein","accounts_in_message":[],"_revision_number":1},{"id":"41849c269a79784e346ec20398fea8b6061292e6","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-11-26 18:32:48.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"4db0b52926c0b0e6689158e37c03d2dec10dcbda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 18:47:07.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caccc1b826584de2b6e30eb8386a194d : SUCCESS in 9m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cfb5b4e9e520499e8ec24dfdd2bead14 : FAILURE in 10m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/a3751d731b024c0d9bb52b1b9f618082 : SUCCESS in 7m 16s","accounts_in_message":[],"_revision_number":2},{"id":"89af7294ce129a0e167af286456189b50578a234","author":{"_account_id":10873,"name":"Juan Antonio Osorio Robles","email":"jaosorior@redhat.com","username":"ejuaoso"},"date":"2019-11-26 18:58:53.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"73110db31ec54147df3b0a44b021cda9450b60a2","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-12-02 17:03:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"09b0852973739e63830040cd97ae959b0b5075e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 17:13:40.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b5e660f4da43ad88826b78e64b6a05 : SUCCESS in 9m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/658f5ae7f6c9414ebf71ed6e32b886de : FAILURE in 9m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d255aa99eeb44fb7ad159718a3367fc4 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":3},{"id":"fde161f519ec5eb7eea8e69779c606d50c16ea7a","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-12-02 17:45:14.000000000","message":"Patch Set 3: Code-Review-1 Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"e49b374083b82632a7405d4c97dd219c74420969","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-12-02 20:12:03.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"e844d761b718d754992f82e045e732c229a41433","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 20:24:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aa69a08d91c4b9097efe7245ca3c0ae : SUCCESS in 9m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e9c63e8f13149deb2a116b0602c9758 : FAILURE in 10m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4c93a0a55e074dd4acfbeaabd77fb028 : SUCCESS in 9m 51s","accounts_in_message":[],"_revision_number":4},{"id":"539acc809de09f4eeeefb35d37f7497471d8d3ea","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-12-02 20:43:00.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5154171821a21f0fbb17aa552f848eac0a785e86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 20:57:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e77dc3cfaf4529b5a2082516764c59 : SUCCESS in 7m 26s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1d32ffd75a53408691f379aa09ab3852 : SUCCESS in 9m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6c67f3b240284e00a2692d284d0cf77e : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":5},{"id":"b64d9a92f47fc18afd1d6303c2849d6159b03394","author":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"date":"2019-12-02 23:55:56.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"7d0edbed7f957c79498728e767209ca390f1804a","author":{"_account_id":14985,"name":"Alex Schultz","email":"aschultz@next-development.com","username":"mwhahaha"},"date":"2019-12-03 00:06:09.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"aa2bcea6805120b606a2cf619ca620e6a2cf2828","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2019-12-03 02:34:34.000000000","message":"Patch Set 5: Code-Review+1\n\nThanks @weshay.","accounts_in_message":[],"_revision_number":5},{"id":"4a92d31913420b59bc05893d708fbd8e506ec535","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-03 14:17:17.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"47dab80629019f599c8c21b2991ad3bd7d622522","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-12-04 09:57:05.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b3b04ace01bc5ae7ec5222c3750d9ce70d0a5e0d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-10 21:52:56.000000000","message":"Patch Set 5:\n\nI believe these numbers are normal for how TripleO has been handling things. It definitely is not normal SemVer, but I think it is what consumers of TripleO have come to expect.","accounts_in_message":[],"_revision_number":5},{"id":"4201766c261501a69cdd7c3a1ecee86ce5573fea","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-12-11 09:27:02.000000000","message":"Patch Set 5:\n\n\u003e I believe these numbers are normal for how TripleO has been\n \u003e handling things. It definitely is not normal SemVer, but I think it\n \u003e is what consumers of TripleO have come to expect.\n\nYes indeed.","accounts_in_message":[],"_revision_number":5},{"id":"2cb062845d8b71885747364cc79bbb05f6faf707","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-11 10:07:50.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"4d75e02f25882e41cca09b2c6ab9aa632d95819b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-12 09:46:42.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"ff1c6cf7e6af7b89f5113c6da9765621f293ad6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 09:47:00.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"78f6f53c7f791eb5d36a8ef9331347d8dcc15a69","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 10:06:01.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a21a4eb62684a978a201735c181e484 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e452102c901a42d38a14e8bdac541b3b : SUCCESS in 10m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/148803f508f6461aa945d5f58a1bfe25 : SUCCESS in 11m 13s","accounts_in_message":[],"_revision_number":5},{"id":"589e2bf3f1aa2c382d07d976813cb70d019aa423","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 10:06:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"02fbce9c9abcda5dce2753bfdb9e60eefadb9d16","revisions":{"1fe13308b82dd3678e6ba623cdf0ddb5604c9205":{"kind":"REWORK","_number":1,"created":"2019-11-26 18:20:55.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/55/696155/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/696155/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/696155/1"}}},"commit":{"parents":[{"commit":"3bb77d72cd61eed924576016af10ee336ec502b6","subject":"Merge \"Release aodhclient 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3bb77d72cd61eed924576016af10ee336ec502b6"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-11-26 18:19:35.000000000","tz":-420},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-11-26 18:19:38.000000000","tz":-420},"subject":"DNM, update release tags for tripleo stein","message":"DNM, update release tags for tripleo stein\n\nBump the release tripleo release tags for\nstein for dependency updates.\n\nChange-Id: I3a9b2ccb88c92a2da92bc1bbdae8380a5a835d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1fe13308b82dd3678e6ba623cdf0ddb5604c9205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1fe13308b82dd3678e6ba623cdf0ddb5604c9205"}]},"branch":"refs/heads/master"},"9f6edec1f37fbe2e444dffd170c060a5772ca7ef":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-26 18:32:48.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/55/696155/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/696155/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/696155/2"}}},"commit":{"parents":[{"commit":"3bb77d72cd61eed924576016af10ee336ec502b6","subject":"Merge \"Release aodhclient 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3bb77d72cd61eed924576016af10ee336ec502b6"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-11-26 18:19:35.000000000","tz":-420},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2019-11-26 18:32:46.000000000","tz":0},"subject":"DNM, update release tags for tripleo stein","message":"DNM, update release tags for tripleo stein\n\nBump the release tripleo release tags for\nstein for dependency updates.\n\nDepends-On: https://review.opendev.org/696157\nChange-Id: I3a9b2ccb88c92a2da92bc1bbdae8380a5a835d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9f6edec1f37fbe2e444dffd170c060a5772ca7ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9f6edec1f37fbe2e444dffd170c060a5772ca7ef"}]},"branch":"refs/heads/master"},"4b2d4b9eb411d61e63410cce89dd286cd8473b9e":{"kind":"REWORK","_number":3,"created":"2019-12-02 17:03:16.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/55/696155/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/696155/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/696155/3"}}},"commit":{"parents":[{"commit":"3bb77d72cd61eed924576016af10ee336ec502b6","subject":"Merge \"Release aodhclient 1.5.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3bb77d72cd61eed924576016af10ee336ec502b6"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-11-26 18:19:35.000000000","tz":-420},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-12-02 17:02:52.000000000","tz":-420},"subject":"update release tags for tripleo stein","message":"update release tags for tripleo stein\n\nBump the release tripleo release tags for\nstein for dependency updates.\n\nDepends-On: https://review.opendev.org/696157\nChange-Id: I3a9b2ccb88c92a2da92bc1bbdae8380a5a835d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b2d4b9eb411d61e63410cce89dd286cd8473b9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4b2d4b9eb411d61e63410cce89dd286cd8473b9e"}]},"branch":"refs/heads/master"},"2aa4f359dc4f0441c9eef52377a2156af0482487":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-12-02 20:12:03.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/55/696155/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/696155/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/696155/4"}}},"commit":{"parents":[{"commit":"f6d69f8c9819796ec9c295fb9a97a787de4ab564","subject":"Merge \"Release openstacksdk 0.38.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f6d69f8c9819796ec9c295fb9a97a787de4ab564"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-11-26 18:19:35.000000000","tz":-420},"committer":{"name":"wes hayutin","email":"whayutin@redhat.com","date":"2019-12-02 20:12:03.000000000","tz":0},"subject":"update release tags for tripleo stein","message":"update release tags for tripleo stein\n\nBump the release tripleo release tags for\nstein for dependency updates.\n\nDepends-On: https://review.opendev.org/696157\nChange-Id: I3a9b2ccb88c92a2da92bc1bbdae8380a5a835d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2aa4f359dc4f0441c9eef52377a2156af0482487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2aa4f359dc4f0441c9eef52377a2156af0482487"}]},"branch":"refs/heads/master"},"02fbce9c9abcda5dce2753bfdb9e60eefadb9d16":{"kind":"REWORK","_number":5,"created":"2019-12-02 20:43:00.000000000","uploader":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"ref":"refs/changes/55/696155/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/55/696155/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/55/696155/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/55/696155/5"}}},"commit":{"parents":[{"commit":"f6d69f8c9819796ec9c295fb9a97a787de4ab564","subject":"Merge \"Release openstacksdk 0.38.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f6d69f8c9819796ec9c295fb9a97a787de4ab564"}]}],"author":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-11-26 18:19:35.000000000","tz":-420},"committer":{"name":"Wes Hayutin","email":"weshayutin@gmail.com","date":"2019-12-02 20:42:44.000000000","tz":-420},"subject":"update release tags for tripleo stein","message":"update release tags for tripleo stein\n\nBump the release tripleo release tags for\nstein for dependency updates.\n\nDepends-On: https://review.opendev.org/696157\nChange-Id: I3a9b2ccb88c92a2da92bc1bbdae8380a5a835d71\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/02fbce9c9abcda5dce2753bfdb9e60eefadb9d16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/02fbce9c9abcda5dce2753bfdb9e60eefadb9d16"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
