)]}'
{"id":"openstack%2Freleases~696173","triplet_id":"openstack%2Freleases~master~Id5009782aec88819e0dc376d98e4bfdc6ad0a8be","project":"openstack/releases","branch":"master","topic":"cinder-eol-driverfixes","hashtags":[],"change_id":"Id5009782aec88819e0dc376d98e4bfdc6ad0a8be","subject":"Cinder project: EOL driverfixes branches","status":"ABANDONED","created":"2019-11-26 19:39:18.000000000","updated":"2020-04-11 20:20:56.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bbbd146bd49ccae279f50528b3f88fbaee09c97c","_number":696173,"virtual_id_number":696173,"owner":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-04-10 19:32:38.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"date":"2019-12-13 18:21:54.000000000","_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-12-13 18:05:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-17 14:45:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2020-01-20 10:11:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-04-10 19:33:42.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-13 18:05:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-01-17 14:45:43.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-01-20 10:11:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-04-10 19:32:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"}],"messages":[{"id":"ec2761049427781fd71ef6a46fd507a9ca570bf4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-26 19:39:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"259e75839acd385825045aed2d0f3c8518219792","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-11-26 19:41:16.000000000","message":"Patch Set 1: Workflow-1\n\nThe email announcing this said it wouldn\u0027t happen until 2 December.  I\u0027m submitting it early just to verify that the files are correct (I haven\u0027t done this before).","accounts_in_message":[],"_revision_number":1},{"id":"7c7340fe6f852e2f3fd9ed67073b9e96884f7b05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 19:50:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/743ef26e1a6d4fe9a9cc242fed7da612 : SUCCESS in 9m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0e47ed176534648b32cd0c4cc361734 : FAILURE in 7m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/cf15482368134ce78f95ac75f0e3735e : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":1},{"id":"269946b15de9d8308758b78c9a2c5943f8fbc97a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-11-27 13:14:15.000000000","message":"Patch Set 1:\n\nHmm, we may need to update some tooling before we can do this. Or probably better given it\u0027s no longer going to be an issue after this point, have some manual intervention to get the driverfixes branches closed out.","accounts_in_message":[],"_revision_number":1},{"id":"d7301631e7e6fc65b96202585f9ae03645050419","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-02 15:26:03.000000000","message":"Patch Set 1:\n\nIn the interim, there\u0027s been some activity on the ML about keeping ocata and pike in EM, so I\u0027ll keep the -W on this for another week.","accounts_in_message":[],"_revision_number":1},{"id":"08f096096d1dfeb54fe1f0c4053f8b57e6c0b9a4","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-13 17:51:36.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54dd5cd762b12c285d5d06265149d36bf5339a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 18:05:02.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1698c6af53742e78b8fa47962ae7221 : SUCCESS in 9m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/745525f1392a4095b9feae3986eeca4f : FAILURE in 7m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9523858b26514755a156f1357581b788 : SUCCESS in 7m 17s","accounts_in_message":[],"_revision_number":2},{"id":"23365f05c1845592577ea8593ad17b20491f663f","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2019-12-13 18:21:54.000000000","message":"Patch Set 2:\n\nAs Sean pointed out on PS1, looks like this request will require some manual intervention.","accounts_in_message":[],"_revision_number":2},{"id":"2b294e92e5a1367e56cb2048ffcd1fbeaa0c6410","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2020-01-17 14:45:43.000000000","message":"Patch Set 2: Code-Review+2\n\nLooks good to me.\n\nWe need to manually trigger this and then recheck and merge this as a \"history\" change, at least that\u0027s how we\u0027ve handled similar in the past","accounts_in_message":[],"_revision_number":2},{"id":"95543bd5fc151ba94d9a4c9076cc39cb14ce6da6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-20 10:11:00.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ea4944613621fa5bd25baa8f3ab4e40fc024a290","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-10 19:32:38.000000000","message":"Patch Set 2:\n\nI have manually pushed these tags. Given the closed state of these branches and that our automation will prevent doing anything with them, I think we can now drop this and ask the infra team to manually delete those branches.","accounts_in_message":[],"_revision_number":2},{"id":"895e6303369c7a66045230a19c9668653f50d295","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-10 19:33:42.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"fdc78d835920dd43e01cae089435ed9da87736a4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-04-11 20:20:56.000000000","message":"Abandoned\n\ndriverfixes branches were manually deleted by Jeremy Stanley on April 10, 2020.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"86d4665098fce206f7c3b33c11c17462c0960afa","revisions":{"4ab3f21cba5e2a78b24189dc71d5bfb5d0fff5b3":{"kind":"REWORK","_number":1,"created":"2019-11-26 19:39:18.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/73/696173/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/696173/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/696173/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/696173/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/696173/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/696173/1"}}},"commit":{"parents":[{"commit":"8568a30288fbe619de74b7989d24efac9485daa5","subject":"Merge \"Release OpenStack-Ansible Rocky\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8568a30288fbe619de74b7989d24efac9485daa5"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-26 19:16:54.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-11-26 19:16:54.000000000","tz":-300},"subject":"Cinder project: EOL some EM branches","message":"Cinder project: EOL some EM branches\n\nAs proposed on the mailing list [0], the following branches are\ntagged EOL at their current HEAD in all Cinder components in\nwhich they exist:\n- driverfixes/mitaka\n- driverfixes/newton\n- stable/ocata\n- stable/pike\n\nThe Cinder team requests that the release team delete the branches\nafter the EOL tags are created.\n\n[0] http://lists.openstack.org/pipermail/openstack-discuss/2019-November/011136.html\n\nChange-Id: Id5009782aec88819e0dc376d98e4bfdc6ad0a8be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ab3f21cba5e2a78b24189dc71d5bfb5d0fff5b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ab3f21cba5e2a78b24189dc71d5bfb5d0fff5b3"}]},"branch":"refs/heads/master"},"86d4665098fce206f7c3b33c11c17462c0960afa":{"kind":"REWORK","_number":2,"created":"2019-12-13 17:51:36.000000000","uploader":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"ref":"refs/changes/73/696173/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/696173/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/696173/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/696173/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/696173/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/696173/2"}}},"commit":{"parents":[{"commit":"301afa7960125f97a5fa8345900b257c41bc1658","subject":"Merge \"Enable check-approval in experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/301afa7960125f97a5fa8345900b257c41bc1658"}]}],"author":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-12-13 17:47:05.000000000","tz":-300},"committer":{"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","date":"2019-12-13 17:47:05.000000000","tz":-300},"subject":"Cinder project: EOL driverfixes branches","message":"Cinder project: EOL driverfixes branches\n\nAs discussed on the mailing list [0,1], the following cinder branches\nare tagged EOL at their current HEAD:\n- driverfixes/mitaka\n- driverfixes/newton\n\nThe Cinder team requests that the release team delete the branches\nafter the EOL tags are created.\n\n[0] http://lists.openstack.org/pipermail/openstack-discuss/2019-November/011136.html\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2019-December/011378.html\n\nChange-Id: Id5009782aec88819e0dc376d98e4bfdc6ad0a8be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/86d4665098fce206f7c3b33c11c17462c0960afa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/86d4665098fce206f7c3b33c11c17462c0960afa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
