)]}'
{"id":"openstack%2Freleases~698491","triplet_id":"openstack%2Freleases~master~I639aa13d9de4f0fd7b1023dc860934fb3429ad13","project":"openstack/releases","branch":"master","topic":"osc-lib","hashtags":[],"change_id":"I639aa13d9de4f0fd7b1023dc860934fb3429ad13","subject":"Release osc-lib for Ussuri milestone 1","status":"MERGED","created":"2019-12-11 15:06:51.000000000","updated":"2019-12-16 10:37:51.000000000","submitted":"2019-12-16 10:37:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698491-1576492671185-8b70566f","meta_rev_id":"565ebbf1eea370936fd682c5bd9765403c3a6284","_number":698491,"virtual_id_number":698491,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-16 10:37:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-12-16 10:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-12-16 10:16:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-12-11 15:27:32.000000000","_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 15:27:32.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2019-12-16 10:16:27.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-12-16 10:37:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b5a7c6471a4240d0166e9681f8e37c5c3a6eeda","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-11 15:06:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f26e685ad4ee0df4f78959530afc91549adda346","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 15:20:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c53cc0b9c5f24dcc823ffa874b6a6203 : SUCCESS in 8m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f6b41d150fa4fd59ae8c532784321c2 : SUCCESS in 4m 34s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/482c15088a284aa7bc337a9fdda60f83 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":1},{"id":"43631765029c530409e6b39a0e56a4c5059ae12c","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-16 10:16:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2b11d170f339983fcd5a19889d7b13a4021d7c2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 10:16:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c11c6c62e98d00f9871696874c985e4ddec0f755","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 10:37:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7ba0bd01f0040a3ba6b7a4ce56a87bb : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2dd91f9a7d84a918d73cfd55a4ef002 : SUCCESS in 8m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0c2d867cd7b44484be1c10fbac97a4c6 : SUCCESS in 9m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3b6cfa77ed14de61a4946cf180f557f884198b00","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 10:37:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bcc551f0f3cdbd02d006c94499da57c9e55c4c24","revisions":{"bcc551f0f3cdbd02d006c94499da57c9e55c4c24":{"kind":"REWORK","_number":1,"created":"2019-12-11 15:06:51.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/91/698491/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/91/698491/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/698491/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/698491/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/91/698491/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/91/698491/1"}}},"commit":{"parents":[{"commit":"b26eea3943260ad8a5f9a50ee264a56b8a56ad20","subject":"Merge \"Release neutron-lib 1.30 for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b26eea3943260ad8a5f9a50ee264a56b8a56ad20"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-11 15:06:36.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-11 15:06:36.000000000","tz":-360},"subject":"Release osc-lib for Ussuri milestone 1","message":"Release osc-lib for Ussuri milestone 1\n\nThis picks up unreleased changes in osc-lib.\nThe cycle-with-intermediary deliverables are good to release frequently\nwhenever there are functional changes to be released. Feel free to -1\nthis patch if there are additional changes that should be included in\nthis release or if it is not a good time to do it.\n\nChange-Id: I639aa13d9de4f0fd7b1023dc860934fb3429ad13\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bcc551f0f3cdbd02d006c94499da57c9e55c4c24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bcc551f0f3cdbd02d006c94499da57c9e55c4c24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
