)]}'
{"id":"openstack%2Freleases~698493","triplet_id":"openstack%2Freleases~master~I1afa11c61545ecc20e0df989841da6ed6c638f78","project":"openstack/releases","branch":"master","topic":"oslo.limit","hashtags":[],"change_id":"I1afa11c61545ecc20e0df989841da6ed6c638f78","subject":"Release oslo.limit for Ussuri milestone 1","status":"MERGED","created":"2019-12-11 15:08:51.000000000","updated":"2019-12-12 09:53:15.000000000","submitted":"2019-12-12 09:53:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698493-1576144395878-b3b91f8b","meta_rev_id":"33d7f5439153a5a0dd83ae1d688cc74a01611b99","_number":698493,"virtual_id_number":698493,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-12 09:53:15.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"recommended":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2019-12-11 16:00:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-12-12 09:36:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-12-12 09:36:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2019-12-11 15:26:39.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-11 15:26:39.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-12-11 16:00:18.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-12-12 09:36:05.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-12-12 09:53:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"18530a7f2493e399626729418179476d579c329c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-11 15:08:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee0b9c4754214c053de95cb4f69bfe33a1b570ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 15:19:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1afbaa043fe4b079d77d87c4ab7efb3 : SUCCESS in 9m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2ba9cdf72673420899aba8e2bc9da9a4 : SUCCESS in 5m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/06bf352e265042b6a19fd08eac097a03 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"7aa5a9bd4cfae25cd1b98379f1239aee19d427f9","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-12-11 16:00:18.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good, we just merged the initial API implementation so this is a good time to release.","accounts_in_message":[],"_revision_number":1},{"id":"a16ff0040d04235a0234784e2d7bcbcd481d9b0f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-12 09:36:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"22811b0bef51d4e585f221de5a9e38c7e6e04254","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 09:36:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"680f30fb2563cabdb60ff57421a2d60b3b8b88eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 09:53:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"459c066ad3cb545c5a17dad6446019c5c155f6c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 09:53:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1fa0bb295ba4866972342faa57933ad : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cefedf9742b420b8bc89a36d54712a0 : SUCCESS in 8m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11fa15515c9c4698a56668d336aaa42b : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8792c0f5d5038a29f03c40f005a4d53a9fd57063","revisions":{"8792c0f5d5038a29f03c40f005a4d53a9fd57063":{"kind":"REWORK","_number":1,"created":"2019-12-11 15:08:51.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/93/698493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/93/698493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/698493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/698493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/93/698493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/93/698493/1"}}},"commit":{"parents":[{"commit":"b26eea3943260ad8a5f9a50ee264a56b8a56ad20","subject":"Merge \"Release neutron-lib 1.30 for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b26eea3943260ad8a5f9a50ee264a56b8a56ad20"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-11 15:08:41.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-11 15:08:41.000000000","tz":-360},"subject":"Release oslo.limit for Ussuri milestone 1","message":"Release oslo.limit for Ussuri milestone 1\n\nThis picks up unreleased changes in oslo.limit.\nThe cycle-with-intermediary deliverables are good to release frequently\nwhenever there are functional changes to be released. Feel free to -1\nthis patch if there are additional changes that should be included in\nthis release or if it is not a good time to do it.\n\nChange-Id: I1afa11c61545ecc20e0df989841da6ed6c638f78\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8792c0f5d5038a29f03c40f005a4d53a9fd57063"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8792c0f5d5038a29f03c40f005a4d53a9fd57063"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
