)]}'
{"id":"openstack%2Freleases~698717","triplet_id":"openstack%2Freleases~master~Id263f0c91f872111fe997e2b8a1088b61033d114","project":"openstack/releases","branch":"master","topic":"diff-gov-releases","hashtags":[],"change_id":"Id263f0c91f872111fe997e2b8a1088b61033d114","subject":"governance/releases consistency check tool","status":"MERGED","created":"2019-12-12 13:29:00.000000000","updated":"2020-02-05 21:38:09.000000000","submitted":"2020-02-05 21:38:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698717-1580938689073-5b512e01","meta_rev_id":"b9af942d0377cd234cef55bef03758533b618d12","_number":698717,"virtual_id_number":698717,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-05 21:38:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-05 21:22:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-01-24 09:37:02.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-05 21:22:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-17 11:48:38.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-01-24 09:37:02.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-05 21:22:18.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-05 21:38:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1bd5a2c0122973a95e700e02c631f9ef4da1c363","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-12 13:29:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6413e712f035cb85002af768d5764f40c6089879","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 13:44:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d8fe16782914853911bb986bab996a8 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a41feaccb604dc4a658bee722bc22be : SUCCESS in 5m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0601736a8c39435e8116c67441b97f10 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/431104149f764328aa4fd06c246e3f53 : SUCCESS in 9m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7829a282cf714c6aa40c441467ec088e : SUCCESS in 5m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b3727dcf2d2542b680eb830e6770b316 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"57729ac32d9c1aaceff0a98e0daf98ddaa08f344","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-19 15:32:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad5081dc8b6eb426a75df799fe1795237590d251","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 15:50:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/433d3b8eee994482b8d51d5bb748cda4 : FAILURE in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d688b5eba024137aea70daac98b6980 : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22c28f91e31947f994577ab3921f7df3 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c2828ce5d004cad943479fcf3f1d850 : SUCCESS in 10m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/04a075ec1be34fd9896f51ed13164eee : SUCCESS in 7m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6160c7ac46eb4ba69675e989c7d03e9e : SUCCESS in 4m 38s","accounts_in_message":[],"_revision_number":2},{"id":"6ddabceb0a7c40dfd6faea32652d03ddc9e3ef97","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-19 17:44:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"48dc32b3f563cbbd6f914622861f36dd46c66cfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 18:02:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/378b330d474d443189fe6d5ab68982a6 : SUCCESS in 4m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b9769a9e1df476e902a01ecd0a97613 : SUCCESS in 4m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d09b92d77e0b44858bae22d23426a898 : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b97a4fc569034a53a83414af8bb1ecff : SUCCESS in 10m 42s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/33d6357cafdd4fdf808337f748c7bec0 : SUCCESS in 7m 57s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1e2d574f5a9343b3934eb21b7a40b001 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":3},{"id":"1771d4bc4dfdb24aec9a18a73fbb142ee95fd702","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-20 16:24:33.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9974bbf09adb471635ce9501b3818936a67d49cf","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2020-01-17 11:48:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)\n\nI like the idea but see inline comments.","accounts_in_message":[],"_revision_number":3},{"id":"8fc6960934a723eecf03c9b8ffd27319b5a308c5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-23 11:39:41.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9b04c8eee8be26dbab43861a6d3f6ff9bf9a8407","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-23 11:42:25.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"a78741090e517d6942fd2f6bfc35d53ad687d460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 11:54:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2aa5a20f9a43442cbcaa07e4cbd41892 : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9785596f5ca5481bb5a8f5f6e8b7a243 : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ae6cd68970494e5f85a57d5a0fb6b86e : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92417b13288944ec8539963e3a5e9483 : SUCCESS in 8m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76172157689d4038ac911fd932ce83a0 : SUCCESS in 5m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/45acf3bb0cde483c979292c8dbf88a49 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":4},{"id":"c0844a75f14ef9d20ed1ad9a9debc9bb880e2527","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-24 09:37:02.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"91003e173c5f74af7b20c5d2138df6606d6d9793","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-05 21:22:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b609df6b4148d357067ba5b3d7f42de2fa0c9ba6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 21:22:46.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"58df0637620b727b4f5d18ccc3565bc2b44691c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 21:38:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c06169431d1c4b769b9d561b916558fa : SUCCESS in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6108db289dfc425f80a093be5925f2b5 : SUCCESS in 5m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/733a1eadb53e4bcbae11fc9aa0a70822 : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/efaef7a87be84107b323ab0242e82bcc : SUCCESS in 10m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cd27da6507a24bbc97e94fd7259dd11b : SUCCESS in 5m 38s","accounts_in_message":[],"_revision_number":4},{"id":"9b3fb914be4fd702db3d202625a799e5e1093087","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-05 21:38:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b54fc8defe0427f3e89059b08a0a8a3e82c551ed","revisions":{"5a9cafd6d532e63976aebb40bea5b0eac848599a":{"kind":"REWORK","_number":1,"created":"2019-12-12 13:29:00.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/17/698717/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/698717/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/698717/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/698717/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/698717/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/698717/1"}}},"commit":{"parents":[{"commit":"cf6e9b7daaa61ff724641d56c821dfa97c4f3de5","subject":"Merge \"Release monasca-common for Ussuri milestone 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf6e9b7daaa61ff724641d56c821dfa97c4f3de5"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-12 13:22:07.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-12 13:22:07.000000000","tz":60},"subject":"[DNM] First draft at a gov/rel consistency check","message":"[DNM] First draft at a gov/rel consistency check\n\nList inconsistencies between the governance and releases repository.\n\nDo not merge, this is posted as an example for now. This tool will be\nrefined as we decide what our policy is for those inconsistencies.\n\nChange-Id: Id263f0c91f872111fe997e2b8a1088b61033d114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a9cafd6d532e63976aebb40bea5b0eac848599a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5a9cafd6d532e63976aebb40bea5b0eac848599a"}]},"branch":"refs/heads/master"},"a0cfb63425972aa8568309fc8afe2fee2fc0ca36":{"kind":"REWORK","_number":2,"created":"2019-12-19 15:32:28.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/17/698717/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/698717/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/698717/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/698717/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/698717/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/698717/2"}}},"commit":{"parents":[{"commit":"cf6e9b7daaa61ff724641d56c821dfa97c4f3de5","subject":"Merge \"Release monasca-common for Ussuri milestone 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf6e9b7daaa61ff724641d56c821dfa97c4f3de5"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-12 13:22:07.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-19 15:27:07.000000000","tz":60},"subject":"governance/releases consistency check tool","message":"governance/releases consistency check tool\n\nIntroduce a new tool to list inconsistencies between the governance\nand releases repository.\n\nSometimes governance defines release-managed deliverables, but the\nrelease team is not handling them. Sometimes, the release management\nteam handles deliverables which are not an official part of OpenStack\nand should probably not be listed on releases.o.o.\n\nThis tool lists both types of inconsistencies, and can be integrated\nto release management process so that they are handled before becoming\nan issue.\n\nNB: Once ready, this tool could be used to replace the \"membership\nfreeze check\" tool which only looked at some of those inconsistencies.\n\nChange-Id: Id263f0c91f872111fe997e2b8a1088b61033d114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0cfb63425972aa8568309fc8afe2fee2fc0ca36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a0cfb63425972aa8568309fc8afe2fee2fc0ca36"}]},"branch":"refs/heads/master"},"8b9b2de9ad348d47587f3377ad1faf0d15c80ff8":{"kind":"REWORK","_number":3,"created":"2019-12-19 17:44:45.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/17/698717/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/698717/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/698717/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/698717/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/698717/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/698717/3"}}},"commit":{"parents":[{"commit":"cf6e9b7daaa61ff724641d56c821dfa97c4f3de5","subject":"Merge \"Release monasca-common for Ussuri milestone 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf6e9b7daaa61ff724641d56c821dfa97c4f3de5"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-12 13:22:07.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-19 17:44:36.000000000","tz":60},"subject":"governance/releases consistency check tool","message":"governance/releases consistency check tool\n\nIntroduce a new tool to list inconsistencies between the governance\nand releases repository.\n\nSometimes governance defines release-managed deliverables, but the\nrelease team is not handling them. Sometimes, the release management\nteam handles deliverables which are not an official part of OpenStack\nand should probably not be listed on releases.o.o.\n\nThis tool lists both types of inconsistencies, and can be integrated\nto release management process so that they are handled before becoming\nan issue.\n\nNB: Once ready, this tool could be used to replace the \"membership\nfreeze check\" tool which only looked at some of those inconsistencies.\n\nChange-Id: Id263f0c91f872111fe997e2b8a1088b61033d114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8b9b2de9ad348d47587f3377ad1faf0d15c80ff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8b9b2de9ad348d47587f3377ad1faf0d15c80ff8"}]},"branch":"refs/heads/master"},"b54fc8defe0427f3e89059b08a0a8a3e82c551ed":{"kind":"REWORK","_number":4,"created":"2020-01-23 11:42:25.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/17/698717/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/698717/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/698717/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/698717/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/698717/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/698717/4"}}},"commit":{"parents":[{"commit":"cf6e9b7daaa61ff724641d56c821dfa97c4f3de5","subject":"Merge \"Release monasca-common for Ussuri milestone 1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/cf6e9b7daaa61ff724641d56c821dfa97c4f3de5"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-12 13:22:07.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-23 11:40:43.000000000","tz":60},"subject":"governance/releases consistency check tool","message":"governance/releases consistency check tool\n\nIntroduce a new tool to list inconsistencies between projects.yaml\nin governance repository and deliverables in the releases repository.\n\nSometimes projects.yaml defines release-managed deliverables, but the\nrelease team is not handling them. Sometimes, the release management\nteam handles deliverables which are not officially produced by an\nOpenStack project team, and should probably not be listed on releases.o.o.\n\nThis tool lists both types of inconsistencies, and can be integrated\nto release management process so that they are handled before becoming\nan issue.\n\nNB: Once ready, this tool could be used to replace the \"membership\nfreeze check\" tool which only looked at some of those inconsistencies.\n\nChange-Id: Id263f0c91f872111fe997e2b8a1088b61033d114\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b54fc8defe0427f3e89059b08a0a8a3e82c551ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b54fc8defe0427f3e89059b08a0a8a3e82c551ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
