)]}'
{"id":"openstack%2Freleases~698736","triplet_id":"openstack%2Freleases~master~I32157fe0a44c8d01c0d01b657a0454be30c35a72","project":"openstack/releases","branch":"master","topic":"release_osa","hashtags":[],"change_id":"I32157fe0a44c8d01c0d01b657a0454be30c35a72","subject":"Release OpenStack-Ansible Train","status":"MERGED","created":"2019-12-12 15:17:47.000000000","updated":"2019-12-16 19:05:31.000000000","submitted":"2019-12-16 19:05:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698736-1576523131586-efdc5bbb","meta_rev_id":"6dba4b1dbc3959998803d77d3d4f2d43b74d35b2","_number":698736,"virtual_id_number":698736,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-16 19:05:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-16 18:52:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-12-16 17:19:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":1,"date":"2019-12-16 17:34:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":1,"date":"2019-12-16 17:40:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-12-16 17:38:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-12-16 18:52:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-16 17:19:44.000000000","updated_by":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"reviewer":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"state":"REVIEWER"},{"updated":"2019-12-16 17:34:21.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-12-16 17:38:07.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-12-16 17:40:46.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-12-16 18:52:01.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-16 19:05:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"333e6d2c9322af798d9548de4bc4a35658634cfe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-12-12 15:17:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b67d0a08b5faed698da76ae207953e67a813dfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 15:36:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d89599b58b84e2e8f718db27c181c8f : SUCCESS in 9m 53s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bac9f5805c2e49599c664690bdba6018 : FAILURE in 6m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/958d140140df49638638516ba38cb808 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f04117889468e5f6dcdde64807968dd6d7dae176","author":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"date":"2019-12-16 17:01:05.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"bc1ef38d2613b9171f5443c2bc72bb56899d9764","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-12-16 17:02:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"a0c89563960c8c6e393a2021db531509e2c01a2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 17:18:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d661ab15805845f886a60eb266d604bc : SUCCESS in 10m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b2a3771b5e3d4f8299d84f9d01cb6dfe : SUCCESS in 5m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3942cd5b7ed54399ad3a6516f59d2f1e : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":1},{"id":"2541d59be19e478a97d4589ae9901d2ae1aa89d2","author":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"date":"2019-12-16 17:19:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1abaab1ab6a20c674e08775019eba4c9e5a32d89","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-12-16 17:34:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d8c40ccc78559c561578e22f97866c82ea7d4071","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-12-16 17:38:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"791e0e52bfe3b0721540c9b1b2e00cc4addb27ed","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-12-16 17:40:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3d4a1a26fd84f9095ffd2a44b3460000bd05348f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-16 18:50:25.000000000","message":"Patch Set 1:\n\nCan you include the branch creation with this? Or submit a follow up patch that we can get through right away with the RC1 cut?","accounts_in_message":[],"_revision_number":1},{"id":"ea24492050bf0eff50247fe2cf79eb40ac77ee8a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-16 18:51:22.000000000","message":"Patch Set 1:\n\nNevermind, missed this at the RC1 creation, so we\u0027ll have to pick that up now. Still needed though.","accounts_in_message":[],"_revision_number":1},{"id":"80f1a2a1f0d09bd11d61596fe78904d5542f34a5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-16 18:52:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nJust ignore me. :]","accounts_in_message":[],"_revision_number":1},{"id":"f2f3b8a89b3413f64b35f295841838ed3577cf27","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 18:52:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eda4b2693097db654ca2476e5462f8ae159ac11a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 19:05:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d711eb3c02b7457ca09814775b34caf7 : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d43553b79198455b8bf84a8cdc859357 : SUCCESS in 8m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/effbc7e199c14cca856df2681e93a92e : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1},{"id":"6b8018d4d2e4e7fb5b4711b0ed4f4d912cfb4bbb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 19:05:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f9530668ad2838117ab4d5347b2639f9d272710","revisions":{"8f9530668ad2838117ab4d5347b2639f9d272710":{"kind":"REWORK","_number":1,"created":"2019-12-12 15:17:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/36/698736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/36/698736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/698736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/698736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/36/698736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/36/698736/1"}}},"commit":{"parents":[{"commit":"9f65e01e40b66be275da5653346d30a13a17572a","subject":"Release OpenStack-Ansible Train RC4","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9f65e01e40b66be275da5653346d30a13a17572a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-12-12 15:17:34.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-12-12 15:17:34.000000000","tz":120},"subject":"Release OpenStack-Ansible Train","message":"Release OpenStack-Ansible Train\n\nChange-Id: I32157fe0a44c8d01c0d01b657a0454be30c35a72\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8f9530668ad2838117ab4d5347b2639f9d272710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/8f9530668ad2838117ab4d5347b2639f9d272710"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
