)]}'
{"id":"openstack%2Freleases~698751","triplet_id":"openstack%2Freleases~master~Ib9d4f66940a631504cc30bb641747ca53debacaf","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ib9d4f66940a631504cc30bb641747ca53debacaf","subject":"[doc]Correcting broken link","status":"ABANDONED","created":"2019-12-12 16:19:30.000000000","updated":"2019-12-16 17:58:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0f7dc6c006af8f9efbbf6d99cc2c1fa117cff226","_number":698751,"virtual_id_number":698751,"owner":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2019-12-12 16:36:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-12 16:36:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c92cd7de9ab8f5d15aabd199dd4843bc0b81d355","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-12 16:19:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9ac9de17df03812b91ee3c8e2275cdc4718c1b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-12 16:36:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c22ee0a7fcd24c4586213aa96d7adf01 : SUCCESS in 11m 30s","accounts_in_message":[],"_revision_number":1},{"id":"0f7dc6c006af8f9efbbf6d99cc2c1fa117cff226","author":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"date":"2019-12-16 17:58:17.000000000","message":"Abandoned\n\nfixed in https://review.opendev.org/#/c/699145/","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6fcf6417b323884106d701518c59bbe4d14ab7c","revisions":{"b6fcf6417b323884106d701518c59bbe4d14ab7c":{"kind":"REWORK","_number":1,"created":"2019-12-12 16:19:30.000000000","uploader":{"_account_id":29313,"name":"Vishal Manchanda","email":"manchandavishal143@gmail.com","username":"vishalmanchanda"},"ref":"refs/changes/51/698751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/698751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/698751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/698751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/698751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/698751/1"}}},"commit":{"parents":[{"commit":"d9e9188a712dc17830f10cba633def4eb861c6a6","subject":"Merge \"update release tags for tripleo stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9e9188a712dc17830f10cba633def4eb861c6a6"}]}],"author":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-12-12 16:13:16.000000000","tz":0},"committer":{"name":"manchandavishal","email":"manchandavishal143@gmail.com","date":"2019-12-12 16:19:28.000000000","tz":0},"subject":"[doc]Correcting broken link","message":"[doc]Correcting broken link\n\nChange-Id: Ib9d4f66940a631504cc30bb641747ca53debacaf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6fcf6417b323884106d701518c59bbe4d14ab7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b6fcf6417b323884106d701518c59bbe4d14ab7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
