)]}'
{"id":"openstack%2Freleases~698845","triplet_id":"openstack%2Freleases~master~I9331bbf87aeb68675afc19f4e4706fbaef2396af","project":"openstack/releases","branch":"master","topic":"release-mistral-ussuri-1","hashtags":[],"change_id":"I9331bbf87aeb68675afc19f4e4706fbaef2396af","subject":"Release Ussuri 1 for Mistral","status":"MERGED","created":"2019-12-13 08:33:09.000000000","updated":"2019-12-13 17:24:07.000000000","submitted":"2019-12-13 17:24:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698845-1576257847451-349857f7","meta_rev_id":"7adb3035298d431eea81bf3b01120a668aa51346","_number":698845,"virtual_id_number":698845,"owner":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-13 17:24:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-13 17:07:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-12-13 10:52:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2019-12-13 17:07:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-13 10:52:04.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-12-13 17:07:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-13 17:24:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f03b82e314ab47022d7ea21cf94290af5ee0f00c","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-13 08:33:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdb1c7fbe386f852b07585ef5c246a3d64e01392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 08:43:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/caa7cf611c264746a8a8bf290a9629d7 : SUCCESS in 9m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a383a834a606451ba4cefc630ad8c75a : SUCCESS in 7m 45s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/86881ecf06ab45ac9c945257df581a26 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"3cf8178a03d5f8bb0e4c14c7351554c447f7dccb","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-13 09:17:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"433d8b8111eb998e49a9a003d5d3affdde2bd701","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-13 09:22:50.000000000","message":"Patch Set 1: Code-Review-1\n\nSince mistral 2.0.0, the mistral-dashboard repository is included in the mistral release... You should add it here (see train/mistral.yaml file for an example).\n\nIf you really want to release mistral-dashboard separately you will have to adjust your deliverable definition in the openstack/governance reference/projects.yaml file, so that mistral-dashboard has its own deliverable.","accounts_in_message":[],"_revision_number":1},{"id":"67beb65d3df319bcad9789b54f54d6d51d69eab3","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-13 09:38:17.000000000","message":"Patch Set 1:\n\nOk, I\u0027ll fix it. Np.","accounts_in_message":[],"_revision_number":1},{"id":"6aafba35510c25f1824d4bdf3fb8df5742add20e","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-13 09:40:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f1ced7abcbc549d63cc13e1f22c9a69567445043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 09:59:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df325c5a8ef1491f91f5ea263296f8c4 : SUCCESS in 9m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e24331306ea44b58645bf442426127e : FAILURE in 7m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/999aa772014a47c09daf78cd4eb48360 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":2},{"id":"e247f18e69b9d170431d21897d8b34f72be9a32d","author":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"date":"2019-12-13 10:09:38.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"706d01c6c8f88b08902d59f0872b59a1dc5f813e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 10:24:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20cfff98ac844fa3b33371ac7474d875 : SUCCESS in 9m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d65e7a17ee55470b8031cb6b8db4311b : SUCCESS in 5m 46s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4f98a0d67d37481cb172ff79bdb0f158 : SUCCESS in 5m 59s","accounts_in_message":[],"_revision_number":3},{"id":"174d3542466ae3118ae6bb964d16b07a0b0f21b3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-13 10:52:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"1afdd428d5f0e97eba4c22358d2aa9b3c3e4308d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-13 17:07:13.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c3463f30b12479deb86b26a31a578cadfb54e2cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:07:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"356973067c761df42b5a6e24e0803c7582f5609c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:24:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"0922e4102168df9e185c680509db928b0604665f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 17:24:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9803d672a0f425e96ba48a341b87d81 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63b9847afaf4430f990e753b3e735cfe : SUCCESS in 11m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40605783dfea4af8b9eb5fb7f52a07aa : SUCCESS in 7m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"2b4c8fc126b9078ebff9fad93b0538fdf1a2567a","revisions":{"298aa981293ceb97c2179803a605887612e8fbd2":{"kind":"REWORK","_number":1,"created":"2019-12-13 08:33:09.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/45/698845/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/45/698845/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/698845/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/698845/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/698845/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/45/698845/1"}}},"commit":{"parents":[{"commit":"d9e9188a712dc17830f10cba633def4eb861c6a6","subject":"Merge \"update release tags for tripleo stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9e9188a712dc17830f10cba633def4eb861c6a6"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-13 08:31:31.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-13 08:31:31.000000000","tz":420},"subject":"Release Ussuri 1 for Mistral","message":"Release Ussuri 1 for Mistral\n\nChange-Id: I9331bbf87aeb68675afc19f4e4706fbaef2396af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/298aa981293ceb97c2179803a605887612e8fbd2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/298aa981293ceb97c2179803a605887612e8fbd2"}]},"branch":"refs/heads/master"},"15b0fc59bd93d3b79b3dd44348ac218349fd56fa":{"kind":"REWORK","_number":2,"created":"2019-12-13 09:40:59.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/45/698845/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/45/698845/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/698845/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/698845/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/698845/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/45/698845/2"}}},"commit":{"parents":[{"commit":"d9e9188a712dc17830f10cba633def4eb861c6a6","subject":"Merge \"update release tags for tripleo stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9e9188a712dc17830f10cba633def4eb861c6a6"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-13 08:31:31.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-13 09:40:43.000000000","tz":420},"subject":"Release Ussuri 1 for Mistral","message":"Release Ussuri 1 for Mistral\n\nChange-Id: I9331bbf87aeb68675afc19f4e4706fbaef2396af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/15b0fc59bd93d3b79b3dd44348ac218349fd56fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/15b0fc59bd93d3b79b3dd44348ac218349fd56fa"}]},"branch":"refs/heads/master"},"2b4c8fc126b9078ebff9fad93b0538fdf1a2567a":{"kind":"REWORK","_number":3,"created":"2019-12-13 10:09:38.000000000","uploader":{"_account_id":8731,"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","username":"rakhmerov"},"ref":"refs/changes/45/698845/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/45/698845/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/698845/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/698845/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/45/698845/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/45/698845/3"}}},"commit":{"parents":[{"commit":"d9e9188a712dc17830f10cba633def4eb861c6a6","subject":"Merge \"update release tags for tripleo stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9e9188a712dc17830f10cba633def4eb861c6a6"}]}],"author":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-13 08:31:31.000000000","tz":420},"committer":{"name":"Renat Akhmerov","email":"renat.akhmerov@gmail.com","date":"2019-12-13 10:09:23.000000000","tz":420},"subject":"Release Ussuri 1 for Mistral","message":"Release Ussuri 1 for Mistral\n\nChange-Id: I9331bbf87aeb68675afc19f4e4706fbaef2396af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b4c8fc126b9078ebff9fad93b0538fdf1a2567a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2b4c8fc126b9078ebff9fad93b0538fdf1a2567a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
