)]}'
{"id":"openstack%2Freleases~698931","triplet_id":"openstack%2Freleases~master~I172f5c0b4785893715e798293bb8ddd658d60b8d","project":"openstack/releases","branch":"master","topic":"exec-on-node","hashtags":[],"change_id":"I172f5c0b4785893715e798293bb8ddd658d60b8d","subject":"Enable check-release-approval on experimental","status":"MERGED","created":"2019-12-13 15:20:17.000000000","updated":"2020-01-27 14:06:23.000000000","submitted":"2020-01-27 14:06:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698931-1580133983143-159bbb0d","meta_rev_id":"d3babd25d2229fee309e48efe1b07b1d624ed863","_number":698931,"virtual_id_number":698931,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-01-27 11:12:57.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-27 14:06:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-27 13:45:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-01-27 13:45:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-20 16:28:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-27 13:45:41.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-27 14:06:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac7940295cd38831fe2db2e77e7b4a2145eb4c54","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-13 15:20:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1117fe5f698ef649ef58f3f85d477570d4c28bfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 15:37:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa6c4e4579642cd9310c2b8a3b00566 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed1f973e3d674467aeb53ebf11523574 : SUCCESS in 6m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/221f7b50462b4ac28dd5491aeb5697f7 : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a298a4b308214d83b31b973ccb71e56c : SUCCESS in 10m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/912a4e18647e4b1bb44bf3984cea4b33 : SUCCESS in 6m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ac73bd7b9af54667b127eab0991f4b42 : SUCCESS in 6m 33s","accounts_in_message":[],"_revision_number":1},{"id":"47dba951185d582879945945ff805ce66237596e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-13 15:44:41.000000000","message":"Patch Set 1: Workflow-1\n\nThis is actually not executing anything, because tox jobs rely on \"hosts:all\" playbooks which are skipped on the executor.\n\nThe job needs to be rewritten (bypassing the default tox job) for this to work.","accounts_in_message":[],"_revision_number":1},{"id":"36bf81ab34687d2f3af53231a6ab55485dce4fab","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-20 16:28:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"77af1ff0066094d383ff54cdedbb7cd76804b720","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-20 16:28:29.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"23f453bd18cdfd1f6d39d1eefa25289bd552172d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-27 10:34:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cb4b97e5eee9db2154aaca95b5d50299c24af5f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 10:46:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a650e11529a4624907f1582c45440ad : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b22a47e019f4fdb96243c324b6a82b5 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0425d05ed85e46d7b291edde8103502e : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e3229c152594a889819157a47d52084 : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/470dc97a0cba4b63be321b17b586fdd8 : SUCCESS in 10m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/481e029caf5b4431a5e22b05aa7e70f0 : SUCCESS in 5m 33s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f41ded58b0d946b98f99ccf516619401 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":2},{"id":"a38d10e76b159f84637fa100bd96212eb3b5620f","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-27 11:12:57.000000000","message":"Patch Set 2:\n\nThis works and can be approved now, so that we start testing it using \"check experimental\" on random release requests.","accounts_in_message":[],"_revision_number":2},{"id":"3df27ef0a5551282803431d77fe91b7652bbf385","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-27 13:45:41.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cb30f5016fed20abc44a71d7216f6d0c92695b62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 13:45:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"55db9d2db35c22911174b2a79f418c4c9292d30e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:06:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9ba93523d5b4922aaf27c6421ecbd16 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c02df4de1846440d8c87d048915851cc : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a0087475636b4d2f89b205e4da4b050b : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5028b9322bcf42e09feaa851ea46ec25 : SUCCESS in 9m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/bb9aa4fbd3284b0a9f6aaae799d33448 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":2},{"id":"a6e82c4f57165ee651c10309ff7910a92707a054","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 14:06:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a56bc4f46cd0bb919b5578ee97dc0fa134d5b893","revisions":{"f9a0d140fc436d66ae321fd54ca4c36ba6409bb4":{"kind":"REWORK","_number":1,"created":"2019-12-13 15:20:17.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/31/698931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/698931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/698931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/698931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/698931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/698931/1"}}},"commit":{"parents":[{"commit":"301afa7960125f97a5fa8345900b257c41bc1658","subject":"Merge \"Enable check-approval in experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/301afa7960125f97a5fa8345900b257c41bc1658"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-13 15:18:47.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-13 15:18:47.000000000","tz":60},"subject":"Execute check_approval directly on Zuul node","message":"Execute check_approval directly on Zuul node\n\ncheck_approval is meant to be light and quick. It should be\nrun on the Zuul node itself.\n\nChange-Id: I172f5c0b4785893715e798293bb8ddd658d60b8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9a0d140fc436d66ae321fd54ca4c36ba6409bb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9a0d140fc436d66ae321fd54ca4c36ba6409bb4"}]},"branch":"refs/heads/master"},"a56bc4f46cd0bb919b5578ee97dc0fa134d5b893":{"kind":"REWORK","_number":2,"created":"2020-01-27 10:34:28.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/31/698931/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/31/698931/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/698931/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/698931/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/31/698931/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/31/698931/2"}}},"commit":{"parents":[{"commit":"301afa7960125f97a5fa8345900b257c41bc1658","subject":"Merge \"Enable check-approval in experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/301afa7960125f97a5fa8345900b257c41bc1658"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-13 15:18:47.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-27 10:32:01.000000000","tz":60},"subject":"Enable check-release-approval on experimental","message":"Enable check-release-approval on experimental\n\nEnable the new run-on-executor check-release-approval job\n(defined in project-config[1]) on the experimental pipeline.\n\n[1] https://review.opendev.org/#/c/701982/\n\nChange-Id: I172f5c0b4785893715e798293bb8ddd658d60b8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a56bc4f46cd0bb919b5578ee97dc0fa134d5b893"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a56bc4f46cd0bb919b5578ee97dc0fa134d5b893"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
