)]}'
{"id":"openstack%2Freleases~698942","triplet_id":"openstack%2Freleases~master~I18034724c0ec4a6a14f6b66808365e69df660dc6","project":"openstack/releases","branch":"master","topic":"neutron-stable-u1","hashtags":[],"change_id":"I18034724c0ec4a6a14f6b66808365e69df660dc6","subject":"Neutron stable releases (Rocky, Stein, Train)","status":"MERGED","created":"2019-12-13 15:55:28.000000000","updated":"2019-12-19 14:42:54.000000000","submitted":"2019-12-19 14:26:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"698942-1576765593077-7a9322a4","meta_rev_id":"e230436fe0141bec00e32a27a8648a0eb2deb54d","_number":698942,"virtual_id_number":698942,"owner":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 14:26:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-13 17:48:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-12-19 14:03:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-12-13 22:09:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-12-19 14:03:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-13 17:48:21.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-13 22:09:40.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2019-12-19 14:03:44.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-12-19 14:26:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6240006fd096ed04fc2ecf7288a81ff37daeaa90","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-13 15:55:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c622c03554d81eed86616bb3e08ad19b27eae02e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-13 16:13:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a8a1ee922af4156902a134a00c28f49 : SUCCESS in 11m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3bcf1dc731464517bd40319f62ab78da : SUCCESS in 11m 58s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/50036f6ecb3d43de839c9d8cf5d8b3e9 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":1},{"id":"c32cff84b143f99693cab9d85ee6a059ade033a7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-13 17:48:21.000000000","message":"Patch Set 1: Code-Review+2\n\nSeems OK. Just need to wait until Monday for stable review.","accounts_in_message":[],"_revision_number":1},{"id":"0461e5ac1c89b17e8896016f19414aff10672ca6","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2019-12-13 22:09:40.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1bab06a53aaf3691b9a09e1fc97c48838874b8de","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-19 10:01:37.000000000","message":"Patch Set 1:\n\n\"Add radvd_user config option\" goes a bit against stable policy, but that ship seems to have sailed ?","accounts_in_message":[],"_revision_number":1},{"id":"a4b8053a883b124e709b91776c5b9b1d41c2f5d7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-19 12:59:32.000000000","message":"Patch Set 1:\n\n\u003e \"Add radvd_user config option\" goes a bit against stable policy,\n \u003e but that ship seems to have sailed ?\n\nThat one caught my attention too, so I had looked closer. It\u0027s a config option added to address a real bug, and the default for the config option appears to make it so not setting it keeps the behavior consistent to what it was before. There is also a good release note included that alerts to the new config option and why it is needed. So it looks like it was done in a way that is consistent with our stable policies.","accounts_in_message":[],"_revision_number":1},{"id":"13414776c5e692fd3dbfb050b26fdbc9651860c3","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-19 14:03:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nSail away","accounts_in_message":[],"_revision_number":1},{"id":"6a3c8f8d1169074aa9a0da968395b9c73848f39d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 14:03:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"81c096b5f97b0278c0b4fda664c0a7d57de4acb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 14:26:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/922a13390d2c4466ba7f0404d52f80a6 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b39865eb9a5740dbb17fcae1f681e329 : SUCCESS in 10m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f4e4cac43d54fca8e68564bee89d0bd : SUCCESS in 14m 43s","accounts_in_message":[],"_revision_number":1},{"id":"74b4bed4ca818e94a6a75d2c3eecbc07f615a8b1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 14:26:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3e9f6fa27f3e1be9d2ccfb72f9703c78bffc090b","author":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"date":"2019-12-19 14:42:54.000000000","message":"Patch Set 1:\n\nNicely summed up by Sean, just to confirm nothing changes if the new configuration option is not set (and default value has same behaviour)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b5fecf3830cf727fd15ff29ce8534d3e2a7f99bd","revisions":{"b5fecf3830cf727fd15ff29ce8534d3e2a7f99bd":{"kind":"REWORK","_number":1,"created":"2019-12-13 15:55:28.000000000","uploader":{"_account_id":21798,"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","username":"bcafarel"},"ref":"refs/changes/42/698942/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/698942/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/698942/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/698942/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/698942/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/698942/1"}}},"commit":{"parents":[{"commit":"301afa7960125f97a5fa8345900b257c41bc1658","subject":"Merge \"Enable check-approval in experimental pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/301afa7960125f97a5fa8345900b257c41bc1658"}]}],"author":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-12-13 15:54:16.000000000","tz":60},"committer":{"name":"Bernard Cafarelli","email":"bcafarel@redhat.com","date":"2019-12-13 15:54:16.000000000","tz":60},"subject":"Neutron stable releases (Rocky, Stein, Train)","message":"Neutron stable releases (Rocky, Stein, Train)\n\nTo go along with Ussuri-1 milestone\n\nChange-Id: I18034724c0ec4a6a14f6b66808365e69df660dc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b5fecf3830cf727fd15ff29ce8534d3e2a7f99bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b5fecf3830cf727fd15ff29ce8534d3e2a7f99bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
