)]}'
{"id":"openstack%2Freleases~699150","triplet_id":"openstack%2Freleases~master~Ic13d6053bce30fe093dfda60c5be4d8f4d5c677b","project":"openstack/releases","branch":"master","topic":"troveclient","hashtags":[],"change_id":"Ic13d6053bce30fe093dfda60c5be4d8f4d5c677b","subject":"Release python-troveclient 3.2.0 for Ussuri","status":"MERGED","created":"2019-12-16 03:37:15.000000000","updated":"2019-12-16 10:37:53.000000000","submitted":"2019-12-16 10:37:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699150-1576492673566-b807434a","meta_rev_id":"b29af114e5e1420fae0cf010e475efe9bf93096a","_number":699150,"virtual_id_number":699150,"owner":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-16 10:37:53.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2019-12-16 10:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2019-12-16 10:18:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-16 10:18:01.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2019-12-16 10:37:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e055bb0e37cd2b6e13561dc3eff5aa971e6ff9a4","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-16 03:37:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f45a4ed47fe9ecc03a2d44101b4b11a329774d02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 03:58:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8aa77f6073c4b37b4ed233bd2bdeb65 : SUCCESS in 8m 02s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e707421231a24545812b6db80be7aa0d : FAILURE in 6m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/024061894b5d4aeeae0f32eb7ce1a4e4 : FAILURE in 4m 36s","accounts_in_message":[],"_revision_number":1},{"id":"9e0393ac6718aecf2ef68ca2fb487a9779dee4ac","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2019-12-16 06:14:02.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0b16877afcd53fd09da8a3acb4163e41209e35ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 06:30:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b6d6d008bb64b4bac188b9f769bcec5 : SUCCESS in 10m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e79fca160e5a4e6fafbd35461f16694d : SUCCESS in 5m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d4f7bf06c8744f6a89240697ae53c0e2 : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":2},{"id":"92d347fdfb1dcee3797c7ee89cc4d447196c6797","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-16 10:18:01.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9c8c41c637492bf1410e46d0334f698b0b96af1c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 10:18:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1b22daa472efa6872178a891b2a298f7d7677312","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 10:37:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"f9210a3f95684293730548e0f5c21501724998b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 10:37:53.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10b237fc204b46c2a3eaa1d000cfa6fe : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29127f979b624827a68149e04a298795 : SUCCESS in 8m 38s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/80d6455cb51b48999b581a91dd7bfe77 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bbed57ccc6adf6ddaff373364c1fd79fb95b7978","revisions":{"b78db38ce204d0e27436fec61e4d06862fb574d6":{"kind":"REWORK","_number":1,"created":"2019-12-16 03:37:15.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/50/699150/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/699150/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/699150/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/699150/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/699150/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/699150/1"}}},"commit":{"parents":[{"commit":"e23a1afbf8b40c46165f37a83147af00bd1b216f","subject":"Merge \"Release hacking 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e23a1afbf8b40c46165f37a83147af00bd1b216f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-16 03:37:04.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-16 03:37:04.000000000","tz":780},"subject":"Release python-troveclient 3.2.0 for Ussuri","message":"Release python-troveclient 3.2.0 for Ussuri\n\nChange-Id: Ic13d6053bce30fe093dfda60c5be4d8f4d5c677b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b78db38ce204d0e27436fec61e4d06862fb574d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b78db38ce204d0e27436fec61e4d06862fb574d6"}]},"branch":"refs/heads/master"},"bbed57ccc6adf6ddaff373364c1fd79fb95b7978":{"kind":"REWORK","_number":2,"created":"2019-12-16 06:14:02.000000000","uploader":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"ref":"refs/changes/50/699150/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/50/699150/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/699150/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/699150/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/50/699150/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/50/699150/2"}}},"commit":{"parents":[{"commit":"e23a1afbf8b40c46165f37a83147af00bd1b216f","subject":"Merge \"Release hacking 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e23a1afbf8b40c46165f37a83147af00bd1b216f"}]}],"author":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-16 03:37:04.000000000","tz":780},"committer":{"name":"Lingxian Kong","email":"anlin.kong@gmail.com","date":"2019-12-16 06:13:59.000000000","tz":0},"subject":"Release python-troveclient 3.2.0 for Ussuri","message":"Release python-troveclient 3.2.0 for Ussuri\n\nChange-Id: Ic13d6053bce30fe093dfda60c5be4d8f4d5c677b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bbed57ccc6adf6ddaff373364c1fd79fb95b7978"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bbed57ccc6adf6ddaff373364c1fd79fb95b7978"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
