)]}'
{"id":"openstack%2Freleases~699469","triplet_id":"openstack%2Freleases~master~Id4e5a88d352e95402d0899deceade6fed8146619","project":"openstack/releases","branch":"master","topic":"check-approvals","hashtags":[],"change_id":"Id4e5a88d352e95402d0899deceade6fed8146619","subject":"Make check_approval not require openstack_governance","status":"MERGED","created":"2019-12-17 17:23:58.000000000","updated":"2020-01-10 14:15:49.000000000","submitted":"2020-01-10 14:15:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699469-1578665749596-f3637170","meta_rev_id":"d2ff049a653484c118034b89da3f1622af0209a6","_number":699469,"virtual_id_number":699469,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-10 14:15:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-17 17:31:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2019-12-20 15:54:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-10 13:57:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-17 17:31:24.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-20 15:54:45.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-10 14:15:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9aa18ec3a723597eb7a2f98bf864bf5b1d4f3a22","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-17 17:23:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"832bf4d8db5d6f842200dd41dabaf0f1904b6cb3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-17 17:31:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"46505ddeeb4024e2db6c82dedf1df6f3b6b2f8ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 17:42:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d765f491c0944bb8a579fd81b61885b5 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e85426dff0e430ea3a7ec1059dd123f : SUCCESS in 7m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7697c950cb2a4de0a500eda49f7cd334 : SUCCESS in 7m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b6ad115c6e7434eabdd18053d508be3 : SUCCESS in 10m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f32abc5c2a2d4e4998f14ed913950852 : SUCCESS in 8m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3273a776f6fc48e6a6cbcf2f48ea8eb2 : SUCCESS in 6m 32s","accounts_in_message":[],"_revision_number":1},{"id":"a8ab07f24ceec925b4052794136d2124a21af1d4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-20 15:54:45.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e895883e707eef6d36873c0c73028012636def13","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-10 13:57:58.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31ee56654a23ec84adbbcce4e143a0acec237cbf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 13:58:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fde7483318f8261cb09d78aafe7de67eddcfec80","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 14:15:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c5d1d310691b903b41a4a714c090ad8d9585cf05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-10 14:15:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edbbe495970c4f1ab9e2840bb44473ce : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e77236e166674a6fb49ab65c38546842 : SUCCESS in 8m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/451d9f59916e4432ba3676db9bab93bc : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b8109ab6ee34f13a9a8f76ecd417752 : SUCCESS in 12m 01s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3d6888f4640341d5be35c324e8b6bc24 : SUCCESS in 8m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"53fba5642b19f46c1383034b67be55deae369cf3","revisions":{"53fba5642b19f46c1383034b67be55deae369cf3":{"kind":"REWORK","_number":1,"created":"2019-12-17 17:23:58.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/69/699469/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/699469/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/699469/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/699469/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/699469/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/699469/1"}}},"commit":{"parents":[{"commit":"f1bb2b5a60bcb1a029552e586d74f88fa709b7b3","subject":"Merge \"Oslo releases for 2019-12-16\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f1bb2b5a60bcb1a029552e586d74f88fa709b7b3"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-17 17:21:29.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-17 17:21:29.000000000","tz":60},"subject":"Make check_approval not require openstack_governance","message":"Make check_approval not require openstack_governance\n\nSince the job will have to run on the Zuul executor, it needs to\nonly rely on stdlib, yaml and requests (which are already installed\non the executor).\n\nChange-Id: Id4e5a88d352e95402d0899deceade6fed8146619\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/53fba5642b19f46c1383034b67be55deae369cf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/53fba5642b19f46c1383034b67be55deae369cf3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
