)]}'
{"id":"openstack%2Freleases~699689","triplet_id":"openstack%2Freleases~master~Ie063c5fd3b6594afa6157d9a5764e3e45eaa7c4a","project":"openstack/releases","branch":"master","topic":"data-alignment","hashtags":[],"change_id":"Ie063c5fd3b6594afa6157d9a5764e3e45eaa7c4a","subject":"mistral-dashboard is released as part of mistral","status":"MERGED","created":"2019-12-18 14:27:05.000000000","updated":"2019-12-22 17:54:47.000000000","submitted":"2019-12-19 10:04:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699689-1576749890588-8ae24aaf","meta_rev_id":"c67a4173bdfe75cababa871ed65de1625ad0c332","_number":699689,"virtual_id_number":699689,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-19 10:04:50.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2019-12-18 22:13:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2019-12-19 09:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2019-12-19 09:52:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-18 22:13:00.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2019-12-19 10:04:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9cec2f26a1b5f4ef8648fb33a3cfb9a31cd40423","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-18 14:27:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"60947531f0df25d1f27919499332aad12c8d8d88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 14:39:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d68da60180ba4306a873d54fb78be990 : SUCCESS in 9m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8f3e858db896490a9cc7194565ce50cc : SUCCESS in 6m 11s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0eec3bfb6b8346868f633aad42cbe450 : SUCCESS in 7m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ea9f37dc086f771083531d49cb7234b97a2929b5","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-18 22:13:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"04959c71ef0e9d9391c853629dc71357f9e7a503","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-19 09:52:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2ab25ab4c12a0bf9ec2122bc560a68a21e3d4457","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 09:52:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dae3f4edc830014d1a264eac663a02c69e9783b9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 10:04:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d921a8076d424a70ba3c95cb25e9c1754191f57e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-19 10:04:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/218d9292ac874caba21ce82258acdbfd : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c85244de8fad454495c4e8abed2c3629 : SUCCESS in 9m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aca675813ea442cd9a55fe82d53ddf38 : SUCCESS in 6m 00s","accounts_in_message":[],"_revision_number":1},{"id":"22af8012f7b6e8295702dee2e57c07aacb7212b3","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-20 10:21:37.000000000","message":"Patch Set 1:\n\nquestion: what is the guideline of deliverable files, a separate deliverable file or a single deliverable file to have multiple repos? Horizon team is using deliverable files in openstack/releases repo to check which deliverables are horizon plugin by checking \"type: horizon-plugin\". mistral-dashboard is usually missed from our list as it is not listed in the section of horizon plugins in releases.openstack.org. I added mistral-dashboard as a separate file in I457f377f841858f591c18ae2c17c0764e523d150, but it was reverted by this change. If this is intended by the mistral team, I am okay, but I am not confident we can always cover mistral-dashboard for horizon plugin wide changes.","accounts_in_message":[],"_revision_number":1},{"id":"9815e8709e527e412de92e98095bb4c9f3ccfea9","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-20 10:31:29.000000000","message":"Patch Set 1:\n\nThe default rule is to use \"deliverables\" to group git repositories that are always released together with the same version number. That said, if mistral-dashboard is the only one doing such bundling and it is causing issues for Horizon, we can totally separate it.\n\nHere the issue is that the deliverable defined in governance did not match what we had in governance[1]. Maybe propose a change there first?\n\n[1] https://opendev.org/openstack/governance/src/branch/master/reference/projects.yaml#L1573","accounts_in_message":[],"_revision_number":1},{"id":"1fdf0f0dcab9cf85aa0e2b31e11667fba980b859","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-20 10:32:45.000000000","message":"Patch Set 1:\n\nI mean, \"deliverable defined in releases did not match what we had in governance\"","accounts_in_message":[],"_revision_number":1},{"id":"26f0eb445de107a10799517df7534d887940523f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2019-12-22 17:54:47.000000000","message":"Patch Set 1:\n\nThe follow-up discussion on IRC is found at http://eavesdrop.openstack.org/irclogs/%23openstack-release/%23openstack-release.2019-12-20.log.html#t2019-12-20T10:33:54","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"451b9ee4bc0b652718bdefad31c6c2cf25d25b14","revisions":{"451b9ee4bc0b652718bdefad31c6c2cf25d25b14":{"kind":"REWORK","_number":1,"created":"2019-12-18 14:27:05.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/89/699689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/89/699689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/699689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/699689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/89/699689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/89/699689/1"}}},"commit":{"parents":[{"commit":"67def614b94e040d43950e5f093e8b0dee922afe","subject":"Release kolla-cli 9.0.0 for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/67def614b94e040d43950e5f093e8b0dee922afe"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-18 14:25:30.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2019-12-18 14:25:30.000000000","tz":60},"subject":"mistral-dashboard is released as part of mistral","message":"mistral-dashboard is released as part of mistral\n\nAccording to governance (and the first release in Ussuri),\nthe openstack/mistral-dashboard repository is released as part\nof the mistral deliverable, so it does not need its extra deliverable\nfile.\n\nChange-Id: Ie063c5fd3b6594afa6157d9a5764e3e45eaa7c4a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/451b9ee4bc0b652718bdefad31c6c2cf25d25b14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/451b9ee4bc0b652718bdefad31c6c2cf25d25b14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
