)]}'
{"id":"openstack%2Freleases~699885","triplet_id":"openstack%2Freleases~master~I0c9f97cb12a429d2e54d94df2b3e110156278bc1","project":"openstack/releases","branch":"master","topic":"pgp","hashtags":[],"change_id":"I0c9f97cb12a429d2e54d94df2b3e110156278bc1","subject":"Add signature links for independent deliverables","status":"MERGED","created":"2019-12-18 22:56:48.000000000","updated":"2020-01-20 11:35:32.000000000","submitted":"2020-01-20 11:35:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699885-1579520132117-0e1c537b","meta_rev_id":"d60faa03fdcbe145824eb66eb9876bbb879b3216","_number":699885,"virtual_id_number":699885,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-12-19 12:56:26.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-20 11:35:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-01-17 14:41:50.000000000","_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":2,"date":"2020-01-20 11:20:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-01-20 10:09:19.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":1,"date":"2020-01-20 11:20:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-17 14:41:50.000000000","updated_by":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"reviewer":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"state":"REVIEWER"},{"updated":"2020-01-20 10:09:19.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-20 11:20:00.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-01-20 11:35:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d00bcd3f8739761ed5b2100edc78a9abc8ad7589","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-18 22:56:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5173d828846336fc36690e8a42e64b48a830b2d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 23:12:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82b110f0887c40258c853a978c91d9b3 : SUCCESS in 11m 00s","accounts_in_message":[],"_revision_number":1},{"id":"0ce3f87ceec52fcdebbf55ec8cb1c841027c3335","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-19 10:27:09.000000000","message":"Patch Set 1:\n\nOne issue here is that it creates a *lot* of 404 links for older independent things (for example, pbr\u003c2). I\u0027m not sure what\u0027s worse... no link at all, or 404 links that make it look like you were hacked.","accounts_in_message":[],"_revision_number":1},{"id":"48444e8a3d4ce182cf737b0942868ca15be9e16f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-19 12:56:26.000000000","message":"Patch Set 1:\n\n\u003e One issue here is that it creates a *lot* of 404 links for older\n \u003e independent things (for example, pbr\u003c2). I\u0027m not sure what\u0027s\n \u003e worse... no link at all, or 404 links that make it look like you\n \u003e were hacked.\n\nYeah, definitely a trade off. At this point, there seemed to be far more deliverables *with* a signature than without, so I ended up leaning this way.\n\nI actually wonder if it would make sense to retroactively generate signatures for all of the older deliverables just so we have consistency across the board.","accounts_in_message":[],"_revision_number":1},{"id":"d4f5cf0f95ead04dcf4d379125f02a3ce0582ec0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2019-12-19 14:02:58.000000000","message":"Patch Set 1:\n\nYes, at this point I\u0027m leaning toward approving this, and either retroactively generate signatures, or somehow mark them as needing to be skipped in their old files.","accounts_in_message":[],"_revision_number":1},{"id":"1cb6f7156f822e1261067234b5f29b48f5fd151d","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2020-01-17 14:41:50.000000000","message":"Patch Set 1: Code-Review+2\n\nBased on [1], currently we have:\n\n $ python3 ~/tmp/link-check.py https://releases.openstack.org/independent.html\n Summary for https://releases.openstack.org/independent.html\n \t   0 Ok signatures\n \t   0 Missing signatures\n\nAfter the change we have: \n\n Summary for \n https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_82b/699885/1/check/openstack-tox-docs/82b110f/docs/independent.html\n \t 482 Ok signatures\n \t 496 Missing signatures\n\nSo it\u0027s pretty close to 50/50 okay vs missing.\n\nI think we should merge this as the now accessible signatures are a good thing and we can, as other have said, reduce the 404\u0027s over time in a bunch of different ways\n\n[1] http://paste.openstack.org/show/788540","accounts_in_message":[],"_revision_number":1},{"id":"869bc0850825a6044828ef71ea15f0aea529f995","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-20 10:09:19.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ec4303eb3c9ed07f1b2a7cb53e06fb4c02c5932b","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-20 11:20:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nOK Let\u0027s do this.\n\nAs a followup I think we should introduce a sig-skip feature that we would add to older deliverable files.","accounts_in_message":[],"_revision_number":1},{"id":"4fe8cf162b8d8e7c6a175e6b5ffae3dc1b3fa292","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 11:20:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a57c9c1ea8fe87553d94ae4f84ae2670cc7aa6f2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 11:35:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bff1cce1089c4bb28d9b658edbf00c6d : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21a9573192604c0695c993dda914c348 : SUCCESS in 9m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b7593f0934055a8f2cff2ea7099ddd8924fad63a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 11:35:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2925d575547205a8d4047268de1cf0fabc222252","revisions":{"2925d575547205a8d4047268de1cf0fabc222252":{"kind":"REWORK","_number":1,"created":"2019-12-18 22:56:48.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/85/699885/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/85/699885/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/699885/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/699885/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/85/699885/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/85/699885/1"}}},"commit":{"parents":[{"commit":"b2a90e3f9119add597abb3c8c6e06f474b12a6c6","subject":"Update process for ussuri-1 weeks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b2a90e3f9119add597abb3c8c6e06f474b12a6c6"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-18 22:54:28.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-18 22:54:28.000000000","tz":-360},"subject":"Add signature links for independent deliverables","message":"Add signature links for independent deliverables\n\nWe have been filtering whether to include pgp links for deliverables if\nthe series was \u003e\u003d \u0027O\u0027, which resulted in filtering out any independent\ndeliverables.\n\nThis updates the check to account for independent deliverables. It also\nmakes the table format consistent for pre- and post-Ocata, just leaving\nthe column blank for series where we did not generate signatures.\n\nChange-Id: I0c9f97cb12a429d2e54d94df2b3e110156278bc1\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2925d575547205a8d4047268de1cf0fabc222252"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2925d575547205a8d4047268de1cf0fabc222252"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
