)]}'
{"id":"openstack%2Freleases~700221","triplet_id":"openstack%2Freleases~master~I84e2cf258e9bc4541b7edde55a22a4be24bdb232","project":"openstack/releases","branch":"master","topic":"announce","hashtags":[],"change_id":"I84e2cf258e9bc4541b7edde55a22a4be24bdb232","subject":"Fix library shortname identification in announce script","status":"MERGED","created":"2019-12-20 18:33:48.000000000","updated":"2020-01-09 17:02:18.000000000","submitted":"2020-01-09 17:02:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"700221-1578589338896-fe3cfc39","meta_rev_id":"ce4c0dd9b39655ffd5aad5595f984ebb85d34140","_number":700221,"virtual_id_number":700221,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-09 17:02:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-01-06 10:09:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":1,"date":"2020-01-06 15:10:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2020-01-09 16:46:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-06 15:10:12.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-09 16:46:46.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-01-09 17:02:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a121928e59813875df38c36378bd8337d2c4d181","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2019-12-20 18:33:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9c4a44e3f0dc956730a6084b6af6eb3ec3aacbbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-20 18:46:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0e106f5bb434ce1ac21878cf2be691b : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3daab190c0834d5f9b12321688c7871d : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/93f53a3eb94b493daa87d7f80584f668 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b03866d55c0d4dee974085e6cc7616e8 : SUCCESS in 8m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7765b9a639db4cf393b99d65301266fc : SUCCESS in 4m 37s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3d63695966d74ea7a5b03847a9beea81 : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e16b7ea6aa0437c21b5f7cb9f1aceba6ec25e3ae","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-06 10:09:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"32553bec0b9b333cf3a054b4b82a8952de05b009","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-06 15:10:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3a5a855545921a2a989132d35110cef4c931538d","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-09 16:46:46.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"7ecc9c80c5a2d57d830fb25d22605535cb378694","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 16:46:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"572c42c01d6372fb237fa66d11add5e03125e008","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 17:02:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ced64c123784d53b384bb46177b2ae8 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a1045bb5252642018a3b915a6ea30d51 : SUCCESS in 6m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5c2ca232a74a4009974b822d286b7835 : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/184bc2a020fe463a8732b11f9244dda6 : SUCCESS in 9m 54s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f94f1f38b15b40f598cebe65a5542aee : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9d96ca9e9973c73015e521c760609c080d349d81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-09 17:02:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bcb61fece97aa03d2ef03b8c94e1bb18ee0e219e","revisions":{"bcb61fece97aa03d2ef03b8c94e1bb18ee0e219e":{"kind":"REWORK","_number":1,"created":"2019-12-20 18:33:48.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/21/700221/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/21/700221/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/700221/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/700221/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/21/700221/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/21/700221/1"}}},"commit":{"parents":[{"commit":"62eef1a6f2b87aa5be07c46a057c9c921a9ca35d","subject":"Release kayobe 7.0.0.0rc1 for Train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/62eef1a6f2b87aa5be07c46a057c9c921a9ca35d"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-20 18:30:26.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2019-12-20 18:33:40.000000000","tz":-360},"subject":"Fix library shortname identification in announce script","message":"Fix library shortname identification in announce script\n\nGit repo origins are now removed in CI jobs, so our detection of project\nshort name from looking at the git remote now fails. This resulted in\nrelease announcements linking to \"null\" instead of the actual project\nlocation [0].\n\nThis updates the script logic to detect if that is the case and fall\nback to getting the repo name from the directory, which should be safe\nin a CI job since there is no manipulation of the clones repo directory\nname.\n\nAlso cleans up an unused variable that was left from a previous\nrefactoring.\n\n[0] http://lists.openstack.org/pipermail/release-announce/2019-December/008306.html\n\nChange-Id: I84e2cf258e9bc4541b7edde55a22a4be24bdb232\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bcb61fece97aa03d2ef03b8c94e1bb18ee0e219e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bcb61fece97aa03d2ef03b8c94e1bb18ee0e219e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
