)]}'
{"id":"openstack%2Freleases~701405","triplet_id":"openstack%2Freleases~master~I291eb02489d7587cd0a8dd9145dd3cf6c667015f","project":"openstack/releases","branch":"master","topic":"cliff-2.18.0","hashtags":[],"change_id":"I291eb02489d7587cd0a8dd9145dd3cf6c667015f","subject":"Release cliff 2.18.0","status":"MERGED","created":"2020-01-07 17:19:18.000000000","updated":"2020-01-08 20:18:22.000000000","submitted":"2020-01-08 20:18:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"701405-1578514702646-1e1bda20","meta_rev_id":"74a0c19b2046e28a56c2b737af914496870a36a6","_number":701405,"virtual_id_number":701405,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-08 20:18:22.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-08 19:57:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-01-08 19:57:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-08 19:57:44.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-08 20:18:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"03df6913024da137ab745779d3ecf94eb159b70d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-07 17:19:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5ef2b984eafe831311b12f243c923c10043a1726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 17:35:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/066f28e7095f4aa68f8beb47a637ab04 : SUCCESS in 9m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d22a6ab96ecc477bad362b0fe72be79d : FAILURE in 4m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/337254f1851e4dc6909dc9019404a417 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":1},{"id":"95a683fd22b165c42419133113d9eab3edc6907c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-07 17:43:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"057bc6cf7926233f26de87ca48487e4a39c70273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 17:55:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585e35295684435cb01000f209cea38e : SUCCESS in 10m 34s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9edae108237340c081312de360a2a898 : SUCCESS in 6m 07s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/979537f53634439ca5fc2c0dbc19c6a6 : SUCCESS in 5m 50s","accounts_in_message":[],"_revision_number":2},{"id":"b7c56bf3c4b1bfca675bfce5a2f06bcb958155f6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-08 19:57:44.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nPushing through since the last release was obviously bad and this is impacting some downstream consumers.","accounts_in_message":[],"_revision_number":2},{"id":"6675855be23a093624de75eb8bccb72e3bafeee7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 19:57:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7682bdbeacdcac47b6f46484b8ca5c99c942114e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 20:18:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ea8f5a818bdba7bc2766c59f4f58f3ecf2360d74","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-08 20:18:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cfc0e9df34144c99eceaed4e0a6ac97 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/228865c26c034a5a87820f28391c7d6c : SUCCESS in 8m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5e13d98e62eb4b7e8efd9fc9e0fd07a1 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5c9bcfdc60e614515c831c0552929cf19dc9a683","revisions":{"396276cb8c43d4419cb4ccba0516558460e8a3f0":{"kind":"REWORK","_number":1,"created":"2020-01-07 17:19:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/701405/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/701405/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/701405/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/701405/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/701405/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/701405/1"}}},"commit":{"parents":[{"commit":"f3935f59d59a908fb7724089608679ac101d5f46","subject":"Merge \"Release os-brick 2.5.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f3935f59d59a908fb7724089608679ac101d5f46"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-07 17:15:46.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-07 17:15:46.000000000","tz":0},"subject":"Release cliff 2.18.0","message":"Release cliff 2.18.0\n\nA mistake was made when releasing cliff 2.17.0. Instead of releasing a\ncommit on master, a commit from an arbitrary review branch was released.\nThis means 2.17.0 is actually 2.11.0 plus some commits instead of 2.16.0\nplus some commits. Release 2.18.0 as one step to resolving this.\n\nChange-Id: I291eb02489d7587cd0a8dd9145dd3cf6c667015f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/396276cb8c43d4419cb4ccba0516558460e8a3f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/396276cb8c43d4419cb4ccba0516558460e8a3f0"}]},"branch":"refs/heads/master"},"5c9bcfdc60e614515c831c0552929cf19dc9a683":{"kind":"REWORK","_number":2,"created":"2020-01-07 17:43:27.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/05/701405/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/701405/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/701405/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/701405/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/701405/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/701405/2"}}},"commit":{"parents":[{"commit":"f3935f59d59a908fb7724089608679ac101d5f46","subject":"Merge \"Release os-brick 2.5.9\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f3935f59d59a908fb7724089608679ac101d5f46"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-01-07 17:15:46.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-07 17:43:23.000000000","tz":-360},"subject":"Release cliff 2.18.0","message":"Release cliff 2.18.0\n\nA mistake was made when releasing cliff 2.17.0. Instead of releasing a\ncommit on master, a commit from an arbitrary review branch was released.\nThis means 2.17.0 is actually 2.11.0 plus some commits instead of 2.16.0\nplus some commits. Release 2.18.0 as one step to resolving this.\n\nChange-Id: I291eb02489d7587cd0a8dd9145dd3cf6c667015f\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5c9bcfdc60e614515c831c0552929cf19dc9a683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5c9bcfdc60e614515c831c0552929cf19dc9a683"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
