)]}'
{"id":"openstack%2Freleases~702098","triplet_id":"openstack%2Freleases~master~Ic2d6a6c13570a006633bb2150ceb741e3e20f24a","project":"openstack/releases","branch":"master","topic":"neutron-lib-2.0.0","hashtags":[],"change_id":"Ic2d6a6c13570a006633bb2150ceb741e3e20f24a","subject":"Release neutron-lib 2.0.0 for Ussuri","status":"MERGED","created":"2020-01-11 19:23:47.000000000","updated":"2020-01-13 19:13:02.000000000","submitted":"2020-01-13 19:13:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702098-1578942782680-bbef5f2b","meta_rev_id":"7dc4ec6a292c0ea74d0fde5527e27f1c8a8076f0","_number":702098,"virtual_id_number":702098,"owner":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-13 19:13:02.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-13 18:59:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2020-01-13 13:54:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":1,"date":"2020-01-13 10:27:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2020-01-13 15:09:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":1,"date":"2020-01-13 14:04:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-01-13 18:59:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"value":0,"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-11 21:09:26.000000000","updated_by":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"reviewer":{"_account_id":4694,"name":"Miguel Lavalle","email":"miguel@mlavalle.com","username":"minsel"},"state":"REVIEWER"},{"updated":"2020-01-11 21:09:32.000000000","updated_by":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"reviewer":{"_account_id":1131,"name":"Brian Haley","email":"haleyb.dev@gmail.com","username":"brian-haley"},"state":"REVIEWER"},{"updated":"2020-01-11 21:09:44.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"REVIEWER"},{"updated":"2020-01-13 10:27:44.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2020-01-13 13:54:05.000000000","updated_by":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"reviewer":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"state":"REVIEWER"},{"updated":"2020-01-13 14:04:28.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-13 15:09:18.000000000","updated_by":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"reviewer":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"state":"REVIEWER"},{"updated":"2020-01-13 18:59:57.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-13 19:13:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ceb51afa068838507481238ae7d25705725a6664","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-11 19:23:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2ed60d96bdc0d72d96282572b889a2d77a19743d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-11 19:31:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd4f73d5fc2a4f15ae641066502046a6 : SUCCESS in 7m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/03a5e7c8dbfa45598d6b8e0ba88dec1f : SUCCESS in 5m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/093c26e47e0f4c3f93ff774e772f05a2 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":1},{"id":"bcbbd5fe4ec8f1d0209ad157e5be7c7b680d9fe7","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-11 23:21:00.000000000","message":"Patch Set 1: Code-Review-1\n\nThis includes dropping py2, so since that is a backwards incompatible change, this needs to be a 2.0.0 release.","accounts_in_message":[],"_revision_number":1},{"id":"24cb90814cf580a63eced2ec89c014a2c719ae31","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-12 11:44:39.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1: Code-Review-1\n \u003e \n \u003e This includes dropping py2, so since that is a backwards incompatible change, this needs to be a 2.0.0 release.\n\nThis new minor release does not drop py2 support, but only adds a patch for an existing problem in 1.31.0. Why do you say this version is doing this?\n\nPlease, make a diff between 1.31.0 and 1.31.1 versions.","accounts_in_message":[],"_revision_number":1},{"id":"8d542000867b3db2eeeab9ebc87e68a9217f1486","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-12 11:46:41.000000000","message":"Patch Set 1: Workflow-1\n\nUfffff you are right!!\n\nSean, thanks for your review!","accounts_in_message":[],"_revision_number":1},{"id":"0af2eebaf47412dcbf193ccbc1cdd47f426aee56","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-12 13:50:45.000000000","message":"Patch Set 1: Code-Review-1\n\nneutron-lib 1.31.0 is part of ussuri cycle, so we don\u0027t necessarily have a new release with a bug fix only. We merged python2 drop (major version bump), so we need to bump a major version. I think 2.0.0 would be fine for the new release.\n\nAnother question is whether we need to block neutron-lib 1.31.0 or not.  I don\u0027t think we need to block neutron-lib 1.31.0. IMHO there is no actual consumer at the moment as a new feature in neutron is expected to use it.\n\nThanks,\nAkihiro","accounts_in_message":[],"_revision_number":1},{"id":"8812c2a9b540df8ceefef4de7e4108e11642090f","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-12 17:47:09.000000000","message":"Patch Set 1:\n\n\u003e neutron-lib 1.31.0 is part of ussuri cycle, so we don\u0027t necessarily\n \u003e have a new release with a bug fix only. We merged python2 drop\n \u003e (major version bump), so we need to bump a major version. I think\n \u003e 2.0.0 would be fine for the new release.\n \u003e \n \u003e Another question is whether we need to block neutron-lib 1.31.0 or\n \u003e not.  I don\u0027t think we need to block neutron-lib 1.31.0. IMHO there\n \u003e is no actual consumer at the moment as a new feature in neutron is\n \u003e expected to use it.\n \u003e \n \u003e Thanks,\n \u003e Akihiro\n\n1.31.0 introduces a bug in the port API definition, preventing from updating the QoS policy in the port. This version should be blocked.","accounts_in_message":[],"_revision_number":1},{"id":"cc681780896a259c800fe49855a1d16ad96c343c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-12 17:52:22.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"9b4f4a56dfc0d526f934edb13014e2deedeebd15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-12 18:02:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54271975a09148cd83512c063b6a9016 : SUCCESS in 8m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e717a692eaec4a9c9c66dfd31f9e51fe : SUCCESS in 5m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d9eb0dbdf8234af3af8d4a21831b95a0 : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":2},{"id":"48cf3630eeebb474ed70b7bdd1f127f93ba8034d","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-12 20:42:07.000000000","message":"Patch Set 2: Code-Review+1\n\nIMO version 2.0.0 is also fine for that as it\u0027s python 3 only now.","accounts_in_message":[],"_revision_number":2},{"id":"43ae77a7f57cf54cbb0d67d4937052e6210d088f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-13 08:08:50.000000000","message":"Patch Set 1:\n\nRodolfo Alonso Hernandez wrote:\n\n \u003e Another question is whether we need to block neutron-lib 1.31.0 or\n \u003e not.  I don\u0027t think we need to block neutron-lib 1.31.0. IMHO there\n \u003e is no actual consumer at the moment as a new feature in neutron is\n \u003e expected to use it.\n \u003e \n \u003e Thanks,\n \u003e Akihiro\n\u003e\n\u003e 1.31.0 introduces a bug in the port API definition, preventing from\n\u003e updating the QoS policy in the port. This version should be blocked.\n\nThanks for clarification. I misunderstood the situation a bit.\nLet\u0027s block it in the global requirements and the neutron repo.","accounts_in_message":[],"_revision_number":1},{"id":"13beb1f405eeeabfdf11b6dfada6afecc9879e3f","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-13 08:09:11.000000000","message":"Patch Set 2: Code-Review+1\n\nneutron-lib 2.0.0 looks good.","accounts_in_message":[],"_revision_number":2},{"id":"97e5745e946d067e5df4df0cfd40fbd0e2272c68","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-01-13 10:17:15.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b37de9e1106a83bd62e710dc732965b3625e082d","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-13 10:26:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"35444766e126bdd624619d659925f675e2cc7db1","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2020-01-13 10:27:11.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"93a8c65c2aa3cf0ffe423cd28190734f0e41545f","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2020-01-13 10:27:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"e07a22ed2201e09bf9aa1f43deafb577260d8111","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 10:44:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ddb9e8dc4374f179ddcf76cc921eb1e : SUCCESS in 9m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f8b2a3391b52449db2fe5b34abc1a903 : SUCCESS in 6m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9ca5c394696a49619a7650438ad19550 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":3},{"id":"1f2ebd45d35e3e8de9d875bbb98d87ba252e5b5e","author":{"_account_id":15554,"name":"Bence Romsics","email":"bence.romsics@gmail.com","username":"ebenrom","status":"inactive contributor"},"date":"2020-01-13 13:54:05.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"285b828e7ce723d42016cc94fd91b4b207f361e5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-13 14:04:28.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"47d3dde4dae778ce385383b77042ea13bdda5493","author":{"_account_id":841,"name":"Akihiro Motoki","email":"amotoki@gmail.com","username":"amotoki"},"date":"2020-01-13 15:09:18.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b50223f71f29a83c117afb176b69673bc1b8e15a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-13 18:59:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7d6476e577b2191492df9b1967de34ddbaef7ae7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 19:00:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"5a670b71eec7d66b01c42808b2cc0fea3a36a9e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 19:13:02.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f6807ad6a6a4bf1bca40e9beef4839e : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bf1989bfbe84192945f80f59b756b8b : SUCCESS in 8m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b4c98507686b42c2abd5190f299233ca : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":3},{"id":"ddd714b7bdf56d699b68de7c0a6d5f997bc60db5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 19:13:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a7317bf34253542545be8fafd9bee7ae6dc69697","revisions":{"74e912d1e26b3fd38e1d8807726ccf6741930087":{"kind":"REWORK","_number":1,"created":"2020-01-11 19:23:47.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/98/702098/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/702098/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/702098/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/702098/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/702098/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/702098/1"}}},"commit":{"parents":[{"commit":"ee2df08000c1101fa83cd39704b5bbb79a79687a","subject":"Merge \"Make check_approval not require openstack_governance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee2df08000c1101fa83cd39704b5bbb79a79687a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-11 19:21:19.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-11 19:21:19.000000000","tz":0},"subject":"Release neutron-lib 1.31.1 for Ussuri","message":"Release neutron-lib 1.31.1 for Ussuri\n\nThis new release contains a fix for the previous one:\n- The new port API extension was redefining the \"qos_policy_id\"\n  parameter, instead of adding the new one \"qos_network_policy_id\".\n\nChange-Id: Ic2d6a6c13570a006633bb2150ceb741e3e20f24a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/74e912d1e26b3fd38e1d8807726ccf6741930087"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/74e912d1e26b3fd38e1d8807726ccf6741930087"}]},"branch":"refs/heads/master"},"a9b6936123b5f64b8409f1be8287514c6f121344":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-12 17:52:22.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/98/702098/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/702098/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/702098/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/702098/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/702098/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/702098/2"}}},"commit":{"parents":[{"commit":"ee2df08000c1101fa83cd39704b5bbb79a79687a","subject":"Merge \"Make check_approval not require openstack_governance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee2df08000c1101fa83cd39704b5bbb79a79687a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-11 19:21:19.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-12 17:47:39.000000000","tz":0},"subject":"Release neutron-lib 2.0.0 for Ussuri","message":"Release neutron-lib 2.0.0 for Ussuri\n\nFeatures:\n- Drops Python 2 support.\n\nFixes:\n- Introduces a fix for a port API extension: the extension was\n  redefining the \"qos_policy_id\" parameter, instead of adding\n  the new one \"qos_network_policy_id\".\n\nChange-Id: Ic2d6a6c13570a006633bb2150ceb741e3e20f24a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a9b6936123b5f64b8409f1be8287514c6f121344"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a9b6936123b5f64b8409f1be8287514c6f121344"}]},"branch":"refs/heads/master"},"a7317bf34253542545be8fafd9bee7ae6dc69697":{"kind":"REWORK","_number":3,"created":"2020-01-13 10:27:11.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/98/702098/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/98/702098/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/702098/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/702098/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/98/702098/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/98/702098/3"}}},"commit":{"parents":[{"commit":"ee2df08000c1101fa83cd39704b5bbb79a79687a","subject":"Merge \"Make check_approval not require openstack_governance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ee2df08000c1101fa83cd39704b5bbb79a79687a"}]}],"author":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-11 19:21:19.000000000","tz":0},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2020-01-13 10:27:09.000000000","tz":0},"subject":"Release neutron-lib 2.0.0 for Ussuri","message":"Release neutron-lib 2.0.0 for Ussuri\n\nFeatures:\n- Drops Python 2 support.\n\nFixes:\n- Introduces a fix for a port API extension: the extension was\n  redefining the \"qos_policy_id\" parameter, instead of adding\n  the new one \"qos_network_policy_id\".\n\nChange-Id: Ic2d6a6c13570a006633bb2150ceb741e3e20f24a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7317bf34253542545be8fafd9bee7ae6dc69697"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a7317bf34253542545be8fafd9bee7ae6dc69697"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
