)]}'
{"id":"openstack%2Freleases~702302","triplet_id":"openstack%2Freleases~master~Ifd777ec06e7fa3e4a3247d66ce3c2f848ca8ba9b","project":"openstack/releases","branch":"master","topic":"osc-lib_2.0.0","hashtags":[],"change_id":"Ifd777ec06e7fa3e4a3247d66ce3c2f848ca8ba9b","subject":"Release osc-lib 2.0.0","status":"MERGED","created":"2020-01-13 22:21:20.000000000","updated":"2020-01-14 16:43:18.000000000","submitted":"2020-01-14 16:43:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702302-1579020198712-c94174cc","meta_rev_id":"24bf702b9b6e0abef185e7f63ebfb7ba56cbd8d5","_number":702302,"virtual_id_number":702302,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-14 16:43:18.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-14 16:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-01-14 16:21:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2020-01-14 16:25:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-01-14 16:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-13 22:25:19.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2020-01-14 16:25:11.000000000","updated_by":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"reviewer":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"state":"REVIEWER"},{"updated":"2020-01-14 16:29:16.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-14 16:43:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d6fbd52cd74d691f76d70b61f22fc6e1738bf2db","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-13 22:21:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"90a235391e114fc494c27f6db5215ad9a7975334","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-13 22:23:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6d3787b8e754a56cee165ab153c9f3f48723eae9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2020-01-13 22:25:19.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks right to me.","accounts_in_message":[],"_revision_number":1},{"id":"84087226788ce63c2243b94b8fae658dcfa3b831","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-13 22:39:30.000000000","message":"Patch Set 1: Workflow-1\n\nHold this until we can pick up https://review.opendev.org/702305 (reno for py2 droppage).","accounts_in_message":[],"_revision_number":1},{"id":"c652eb4cf8351442c8944de05ee155ed806fe979","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 23:00:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/180f19b61fb244a48e5b0f34a060142c : SUCCESS in 8m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0142e9ae89504285bcf515bedd3fc31c : SUCCESS in 7m 14s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d03f1ea718d24b30ad019906b718bba2 : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":1},{"id":"1108f9b56833f9b8b89b9c4b15ce4fad98f57fde","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-14 14:35:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"61f7f10ec97f2d5c6d60c094a5037640b84450ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 14:50:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4e04413d0344419beff52a81cc01763 : SUCCESS in 9m 39s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1720556127ac40cdba6c6d470043be86 : SUCCESS in 6m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/0dffc7052c134b3f94ce26969fcb0012 : FAILURE in 6m 27s","accounts_in_message":[],"_revision_number":2},{"id":"3d181ef7f9f8536105723c3b2d5c4281e10d1eb3","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-14 15:02:17.000000000","message":"Patch Set 2:\n\nrecheck\n\nUnable to reproduce locally. Looks like some issue fetching the governance data.","accounts_in_message":[],"_revision_number":2},{"id":"d8facc32c98460993dda99504bca322003866917","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 15:44:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c6162f75f3548b4adc5bedbb22b7217 : SUCCESS in 9m 48s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5719a926819643e787954247b24f0ac0 : SUCCESS in 7m 32s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/50e0f9e705b94af68743de8ea69b6216 : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":2},{"id":"7c6ece90c943e244994b83fa1aace9af07e3abf5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-14 16:21:58.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e132b9d9969dbc50a0a9ac0a2a0861bbc5e59e78","author":{"_account_id":970,"name":"Dean Troyer","email":"dtroyer@gmail.com","username":"dtroyer"},"date":"2020-01-14 16:25:11.000000000","message":"Patch Set 2: Code-Review+1\n\nThanks Eric","accounts_in_message":[],"_revision_number":2},{"id":"c84d7591595f04fa60d856b8c38587f8626f3773","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-14 16:29:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9b22dfaf7bfd31023c5e84e7aeb080fbcb617ea6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 16:29:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e78de37b1ac5c1934ee26ff10e53feb42ab9b4b4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 16:43:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"e35b4d72a2954f5096c8b0fe3f8f0882bbd13188","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-14 16:43:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/867760c75c904341a0dffcdab0be09f9 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a69d11625e743c48db81d189d9d2e21 : SUCCESS in 9m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e49fb76700964019afcb41cbb32fceca : SUCCESS in 6m 31s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3efbf7e2db8e2478e1cf34b2cc5608dfda8752c5","revisions":{"f0035becdcb60ee99947c993c46b91a973b25582":{"kind":"REWORK","_number":1,"created":"2020-01-13 22:21:20.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/02/702302/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/02/702302/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/702302/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/702302/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/702302/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/02/702302/1"}}},"commit":{"parents":[{"commit":"bdd5e522e7a33cd6058f8031b9301867c4eb6095","subject":"Merge \"Address review comments for make_missing_releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bdd5e522e7a33cd6058f8031b9301867c4eb6095"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2020-01-13 22:19:42.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2020-01-13 22:19:42.000000000","tz":-360},"subject":"Release osc-lib 2.0.0","message":"Release osc-lib 2.0.0\n\nThis release picks up [1], which is a feature; but also is the first\nrelease since dropping py2 support (via [2]), hence the major bump.\n\n[1] If700daf8a38b1c426bcfcb5ac37202e8aa0178e5\n[2] Iff873ae15f865aa8740760a3e904a96d22e553e8\n\nChange-Id: Ifd777ec06e7fa3e4a3247d66ce3c2f848ca8ba9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0035becdcb60ee99947c993c46b91a973b25582"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f0035becdcb60ee99947c993c46b91a973b25582"}]},"branch":"refs/heads/master"},"3efbf7e2db8e2478e1cf34b2cc5608dfda8752c5":{"kind":"REWORK","_number":2,"created":"2020-01-14 14:35:12.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/02/702302/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/02/702302/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/702302/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/702302/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/02/702302/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/02/702302/2"}}},"commit":{"parents":[{"commit":"bdd5e522e7a33cd6058f8031b9301867c4eb6095","subject":"Merge \"Address review comments for make_missing_releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bdd5e522e7a33cd6058f8031b9301867c4eb6095"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2020-01-13 22:19:42.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-01-14 14:35:01.000000000","tz":-360},"subject":"Release osc-lib 2.0.0","message":"Release osc-lib 2.0.0\n\nThis release picks up [1], which is a feature; but also is the first\nrelease since dropping py2 support (via [2]), hence the major bump.\n\n[1] If700daf8a38b1c426bcfcb5ac37202e8aa0178e5\n[2] Iff873ae15f865aa8740760a3e904a96d22e553e8\n\nChange-Id: Ifd777ec06e7fa3e4a3247d66ce3c2f848ca8ba9b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3efbf7e2db8e2478e1cf34b2cc5608dfda8752c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3efbf7e2db8e2478e1cf34b2cc5608dfda8752c5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
