)]}'
{"id":"openstack%2Freleases~702652","triplet_id":"openstack%2Freleases~master~Iba9e61c430c6bf31c9761c967dcd52a6d1a7faec","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Iba9e61c430c6bf31c9761c967dcd52a6d1a7faec","subject":"os-traits: release 2.2.0","status":"MERGED","created":"2020-01-15 12:11:33.000000000","updated":"2020-01-15 21:24:28.000000000","submitted":"2020-01-15 21:24:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702652-1579123468278-bbef5f2b","meta_rev_id":"9e983c95f56e53b617a11e3f5eb5621b99c2aaad","_number":702652,"virtual_id_number":702652,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":18081,"name":"Tetsuo NAKAMURA","email":"t-nakamura@bit-isle.co.jp","username":"nakamura"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-15 21:24:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-01-15 21:11:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":18081,"name":"Tetsuo NAKAMURA","email":"t-nakamura@bit-isle.co.jp","username":"nakamura"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-01-15 21:11:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":18081,"name":"Tetsuo NAKAMURA","email":"t-nakamura@bit-isle.co.jp","username":"nakamura"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":18081,"name":"Tetsuo NAKAMURA","email":"t-nakamura@bit-isle.co.jp","username":"nakamura"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":18081,"name":"Tetsuo NAKAMURA","email":"t-nakamura@bit-isle.co.jp","username":"nakamura"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-15 20:38:53.000000000","updated_by":{"_account_id":18081,"name":"Tetsuo NAKAMURA","email":"t-nakamura@bit-isle.co.jp","username":"nakamura"},"reviewer":{"_account_id":18081,"name":"Tetsuo NAKAMURA","email":"t-nakamura@bit-isle.co.jp","username":"nakamura"},"state":"REVIEWER"},{"updated":"2020-01-15 20:43:47.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-01-15 21:11:02.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-15 21:24:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"776aa88a2b73fffd33ed83e7f48e9d21b4b9687d","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-01-15 12:11:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b918e870fb3de20e8eb53182c5c3ba77f768e84f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 12:27:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2493dcbe6dd41ddbf405fd323747715 : SUCCESS in 10m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4b374d0a74964d72a60612cf12662f7e : FAILURE in 5m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/8f48b415fd7b4b41884aeb2527c0ffe5 : SUCCESS in 5m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d99fa5a6d29f4e162aa597d60bfc797fac0251a5","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-01-15 13:17:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a1cfb62ad3efded866f1883f102bf60ee9299614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 13:26:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e931a31646748c8bad1c5f4861b7554 : SUCCESS in 8m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/feeecfd1493c4aea9f05ed9880d2a690 : SUCCESS in 6m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2248ed9cdcb44c3c9c1f8bf4288366d8 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e8198307bbda1faf4947a584e55b1a5c1d3e9e41","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-15 20:39:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"eab0032c7be11d45d6ddad6b176c4f0d4d21cc13","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-15 20:43:47.000000000","message":"Patch Set 2: Code-Review-1\n\nLet\u0027s update to pick up bdc942a27248eeb0ad9b674f099d671e88792d54 (I6dc00248b54f3a8a5d5dae284d2182df5ba550ab)","accounts_in_message":[],"_revision_number":2},{"id":"5a5c885562e5e46ee3a5ed2c59c6940829e539c4","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-15 20:45:09.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"ae3c238d9c8112594874c1d96217f2adba995b34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 20:58:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c245f1a0cf084c26b2b809501b7e3e3d : SUCCESS in 9m 11s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0949dbc0bb4e46a7b8e6b60232b15d6d : SUCCESS in 5m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/335b73fa8ab74c4397d0ad24c3d99a09 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a8c4a5c23dba43b7a4a478a167b348a1c36d1412","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-15 21:11:02.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"b703dab8d1dfa2d4d86948ad878d62054a9fdedc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 21:11:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"0a8b6b7a4e8194acac16d8ba2435831c380e7304","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 21:24:28.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28815ed37d73415baeadd07c95cd3afb : SUCCESS in 3m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c89a396467ea4484a99d5f7106c376fb : SUCCESS in 8m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/259f37b22f9e46da963df543c1fa029e : SUCCESS in 4m 59s","accounts_in_message":[],"_revision_number":3},{"id":"9e983c95f56e53b617a11e3f5eb5621b99c2aaad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 21:24:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f73474b93e9f2e493a34e4b2a3c45a3b2bf2e359","revisions":{"5ade4efab8524484cd3d84fec01b25825f771628":{"kind":"REWORK","_number":1,"created":"2020-01-15 12:11:33.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/52/702652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/52/702652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/702652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/702652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/702652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/52/702652/1"}}},"commit":{"parents":[{"commit":"28196ec86d9e15c5824636125b2df4b8dfcbf09c","subject":"Merge \"Release osc-lib 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28196ec86d9e15c5824636125b2df4b8dfcbf09c"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-01-15 12:08:34.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-01-15 12:11:31.000000000","tz":0},"subject":"os-traits: release 2.2.0","message":"os-traits: release 2.2.0\n\nThis adds a single trait, COMPUTE_RESCUE_BFV:\n\nc37f67a Add COMPUTE_RESCUE_BFV trait\n\nChange-Id: Iba9e61c430c6bf31c9761c967dcd52a6d1a7faec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ade4efab8524484cd3d84fec01b25825f771628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5ade4efab8524484cd3d84fec01b25825f771628"}]},"branch":"refs/heads/master"},"52fb2f078296528a2de130c3a025ae376e3de504":{"kind":"REWORK","_number":2,"created":"2020-01-15 13:17:09.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/52/702652/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/52/702652/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/702652/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/702652/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/702652/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/52/702652/2"}}},"commit":{"parents":[{"commit":"28196ec86d9e15c5824636125b2df4b8dfcbf09c","subject":"Merge \"Release osc-lib 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28196ec86d9e15c5824636125b2df4b8dfcbf09c"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-01-15 12:08:34.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-01-15 13:17:03.000000000","tz":0},"subject":"os-traits: release 2.2.0","message":"os-traits: release 2.2.0\n\nThis adds a single trait, COMPUTE_RESCUE_BFV:\n\nc37f67a Add COMPUTE_RESCUE_BFV trait\n\nChange-Id: Iba9e61c430c6bf31c9761c967dcd52a6d1a7faec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/52fb2f078296528a2de130c3a025ae376e3de504"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/52fb2f078296528a2de130c3a025ae376e3de504"}]},"branch":"refs/heads/master"},"f73474b93e9f2e493a34e4b2a3c45a3b2bf2e359":{"kind":"REWORK","_number":3,"created":"2020-01-15 20:45:09.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/52/702652/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/52/702652/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/702652/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/702652/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/702652/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/52/702652/3"}}},"commit":{"parents":[{"commit":"28196ec86d9e15c5824636125b2df4b8dfcbf09c","subject":"Merge \"Release osc-lib 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28196ec86d9e15c5824636125b2df4b8dfcbf09c"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-01-15 12:08:34.000000000","tz":0},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2020-01-15 20:45:07.000000000","tz":0},"subject":"os-traits: release 2.2.0","message":"os-traits: release 2.2.0\n\nThis adds COMPUTE_RESCUE_BFV and COMPUTE_ACCELERATORS:\n\nc37f67a Add COMPUTE_RESCUE_BFV trait\nbdc942a Add a trait that compute manager can handle accelerator requests.\n\nChange-Id: Iba9e61c430c6bf31c9761c967dcd52a6d1a7faec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f73474b93e9f2e493a34e4b2a3c45a3b2bf2e359"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f73474b93e9f2e493a34e4b2a3c45a3b2bf2e359"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
