)]}'
{"id":"openstack%2Freleases~702917","triplet_id":"openstack%2Freleases~master~I5c80a42fc61f2af7f18510f8275e25693b36e09f","project":"openstack/releases","branch":"master","topic":"sphinx-feature-classification_1.0.0","hashtags":[],"change_id":"I5c80a42fc61f2af7f18510f8275e25693b36e09f","subject":"Release sphinx-feature-classification 1.0.0","status":"MERGED","created":"2020-01-16 16:33:07.000000000","updated":"2020-01-16 17:35:07.000000000","submitted":"2020-01-16 17:35:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702917-1579196107504-bbef5f2b","meta_rev_id":"13ac3423e0e5bedc75ed16a31db7596c2c15e850","_number":702917,"virtual_id_number":702917,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 17:35:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"recommended":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":1,"date":"2020-01-16 16:38:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-01-16 17:22:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2020-01-16 17:22:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 16:38:05.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2020-01-16 17:22:08.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-01-16 17:35:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d52298d04c17f9761507214d0e008feea906304e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-01-16 16:33:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff552408f78df6750747c725e8fef407f330a547","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-01-16 16:38:05.000000000","message":"Patch Set 1: Code-Review+1\n\nPTL +1","accounts_in_message":[],"_revision_number":1},{"id":"a92f676bd48a4004f714e57d9e721eee8ec3064c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 16:48:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b656726022084ea5b081aec25819b069 : SUCCESS in 9m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/dcfc354d3b5e4ec7936cd3c35962ba5b : SUCCESS in 5m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1f552f0bd10d4e36aea3cd61c033e4b3 : SUCCESS in 5m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3996f47333a83a4e429b76c33d431f46e7819250","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-01-16 17:22:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f1ae815b9466e94eab9ca3d70615f772b9c4dfa5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 17:24:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ede28cbe2909d6cb0b635a41e2c352dd2052cb42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 17:35:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/affba6ea06104214927fcee26d7118d5 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96c347dfac0349528007725bdfdd1e47 : SUCCESS in 9m 49s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f06adcbef2fa4324978cfede691b9317 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":1},{"id":"8f075c4f131f925a26957ee9448437b26100cbca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 17:35:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c8b9716ccfb4ec10a57d58f2f1d7c339fb01912a","revisions":{"c8b9716ccfb4ec10a57d58f2f1d7c339fb01912a":{"kind":"REWORK","_number":1,"created":"2020-01-16 16:33:07.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/17/702917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/17/702917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/702917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/702917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/17/702917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/17/702917/1"}}},"commit":{"parents":[{"commit":"ec2635b926e5caada7bab699e6f3420994c1c253","subject":"Merge \"os-traits: release 2.2.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ec2635b926e5caada7bab699e6f3420994c1c253"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2020-01-16 16:31:16.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2020-01-16 16:31:16.000000000","tz":-360},"subject":"Release sphinx-feature-classification 1.0.0","message":"Release sphinx-feature-classification 1.0.0\n\nPick up support for feature.api [1].\n\nMajor bump as this is the first release since dropping py2 support [2].\n\n[1] I36bde39742f7f3ee60ee10240d836c7083cfa0fd\n[2] I5ab48fe70e49d104290f1c01014e4aa6420e300f\n\nChange-Id: I5c80a42fc61f2af7f18510f8275e25693b36e09f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c8b9716ccfb4ec10a57d58f2f1d7c339fb01912a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c8b9716ccfb4ec10a57d58f2f1d7c339fb01912a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
