)]}'
{"id":"openstack%2Freleases~703395","triplet_id":"openstack%2Freleases~master~I83e64de23ae75f5abedc6219a07a8cfdd661687d","project":"openstack/releases","branch":"master","topic":"feature/new_tools","hashtags":[],"change_id":"I83e64de23ae75f5abedc6219a07a8cfdd661687d","subject":"Add docs for tools/make_missing_releases.sh","status":"MERGED","created":"2020-01-20 13:53:15.000000000","updated":"2020-01-23 15:06:38.000000000","submitted":"2020-01-23 15:06:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"703395-1579791998544-46a14a57","meta_rev_id":"f2cb5752d17670a2cef27c0fd09293dc2402495c","_number":703395,"virtual_id_number":703395,"owner":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-23 15:06:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-01-23 12:29:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-23 12:29:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2020-01-23 12:29:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-22 10:39:13.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-01-22 15:17:53.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-01-23 12:29:46.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-01-23 15:06:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"346f323e2cb6bc3a341fe436d0de316d4c16cdff","author":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"date":"2020-01-20 13:53:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e767a3d452e3036098050cd7b428f3d26d9857e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-20 14:25:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12b2897cf37e42d5862ea47dadba597b : SUCCESS in 11m 16s","accounts_in_message":[],"_revision_number":1},{"id":"f8ce30fc0f444c27c379545904ba011ad1d17f64","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-22 10:39:13.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nHello,\n\nSee my inline comment","accounts_in_message":[],"_revision_number":1},{"id":"284272344fe659df37ef200ba56662fbb098c129","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-22 15:17:34.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"01434b5d27e460c6ea8a4b2de5e13b5fa84b02c9","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-01-22 15:17:53.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cfe680f5582664dc67f117540403b2835c29b5a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 15:29:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbbfb095f5d046e5b6079c3ce4d4f478 : SUCCESS in 9m 23s","accounts_in_message":[],"_revision_number":2},{"id":"f80988dc755e45b49829c6ef4ffc33e57aff8ef2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-23 11:05:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8dc38ee98211b6e9683276baf5f779fdcd5b3411","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-23 11:15:37.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"d08d901d08df3ff365e7caae4c6f171f56cd31dd","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-23 11:15:39.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"feb387bf663ee76b4eb490155f5c298e4e8cdab7","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-23 12:29:33.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1014a67b70b0ea6119c83805e4709c28375bdc1a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-01-23 12:29:46.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9abc21a2ce61309d8b7439cadac4e692e2ee5ffc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 12:42:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39ee2fd5b8d54d9794c3d4c31248663f : SUCCESS in 10m 06s","accounts_in_message":[],"_revision_number":3},{"id":"8b4d7deb14bff3e3adac7e473eab310e4b385a72","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 14:49:58.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"19cd748cc562f3436efa74d7be8fdc4810b2241b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 15:06:38.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4ad7f1f5e8b4ab4bc9d01b8052207bc : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37c0aa1bd6804a778d5311b186576ec8 : SUCCESS in 11m 22s","accounts_in_message":[],"_revision_number":3},{"id":"13195bb617b183e5cd721808f792338e3244fe49","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-23 15:06:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c582feffc03b8dc3cc81471f3c0b11d70147ce6c","revisions":{"7ecf93f2dd49a19067ba0a23c62689ea9df49a01":{"kind":"REWORK","_number":1,"created":"2020-01-20 13:53:15.000000000","uploader":{"_account_id":12898,"name":"Tony Breeds","email":"tony@bakeyournoodle.com","username":"tonyb"},"ref":"refs/changes/95/703395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/703395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/703395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/703395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/703395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/703395/1"}}},"commit":{"parents":[{"commit":"276f90bde28a7c40b5b3e6dd08bb3737713332c8","subject":"Add tools/bulk_review.sh and get-contacts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/276f90bde28a7c40b5b3e6dd08bb3737713332c8"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2020-01-20 13:45:32.000000000","tz":60},"committer":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2020-01-20 13:48:18.000000000","tz":60},"subject":"Add docs for tools/make_missing_releases.sh","message":"Add docs for tools/make_missing_releases.sh\n\nChange-Id: I83e64de23ae75f5abedc6219a07a8cfdd661687d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ecf93f2dd49a19067ba0a23c62689ea9df49a01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7ecf93f2dd49a19067ba0a23c62689ea9df49a01"}]},"branch":"refs/heads/master"},"70e6e0305df63d3eb40e7c3f2784a36dd45cbee7":{"kind":"REWORK","_number":2,"created":"2020-01-22 15:17:34.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/95/703395/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/703395/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/703395/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/703395/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/703395/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/703395/2"}}},"commit":{"parents":[{"commit":"276f90bde28a7c40b5b3e6dd08bb3737713332c8","subject":"Add tools/bulk_review.sh and get-contacts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/276f90bde28a7c40b5b3e6dd08bb3737713332c8"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2020-01-20 13:45:32.000000000","tz":60},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2020-01-22 15:17:32.000000000","tz":0},"subject":"Add docs for tools/make_missing_releases.sh","message":"Add docs for tools/make_missing_releases.sh\n\nChange-Id: I83e64de23ae75f5abedc6219a07a8cfdd661687d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/70e6e0305df63d3eb40e7c3f2784a36dd45cbee7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/70e6e0305df63d3eb40e7c3f2784a36dd45cbee7"}]},"branch":"refs/heads/master"},"c582feffc03b8dc3cc81471f3c0b11d70147ce6c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-01-23 12:29:33.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/95/703395/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/95/703395/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/703395/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/703395/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/95/703395/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/95/703395/3"}}},"commit":{"parents":[{"commit":"60e786d7248044fad3db33ed401f69c7d0607e61","subject":"Add tools/bulk_review.sh and get-contacts","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/60e786d7248044fad3db33ed401f69c7d0607e61"}]}],"author":{"name":"Tony Breeds","email":"tony@bakeyournoodle.com","date":"2020-01-20 13:45:32.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-01-23 12:29:33.000000000","tz":0},"subject":"Add docs for tools/make_missing_releases.sh","message":"Add docs for tools/make_missing_releases.sh\n\nChange-Id: I83e64de23ae75f5abedc6219a07a8cfdd661687d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c582feffc03b8dc3cc81471f3c0b11d70147ce6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c582feffc03b8dc3cc81471f3c0b11d70147ce6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
