)]}'
{"id":"openstack%2Freleases~704358","triplet_id":"openstack%2Freleases~master~I6b128ff0d22ce13e1245a4c55ce608142564207d","project":"openstack/releases","branch":"master","topic":"ceilometer-stable","hashtags":[],"change_id":"I6b128ff0d22ce13e1245a4c55ce608142564207d","subject":"Release Ceilometer stable branches","status":"MERGED","created":"2020-01-27 16:59:10.000000000","updated":"2020-02-27 12:34:47.000000000","submitted":"2020-02-27 12:34:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704358-1582806887142-d353a4dd","meta_rev_id":"e654eebe692cdae0b8ae1d8ed6d5052a9a57e5eb","_number":704358,"virtual_id_number":704358,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-27 12:34:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-26 21:22:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-27 12:09:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-02-26 09:16:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-27 12:09:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-02-26 09:17:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-27 21:52:59.000000000","updated_by":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"reviewer":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"state":"REVIEWER"},{"updated":"2020-02-19 10:30:47.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-02-26 09:16:07.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-02-26 21:22:28.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-27 12:09:17.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-27 12:34:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ca6089974b90bc005a3d500fef40273fbea9ddd3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-27 16:59:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48fbd05eedd74712c14c6662adf51fe281e51d06","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-27 17:09:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fff154905e84d969e48e565804185de : SUCCESS in 8m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/eaa64a523b3043f68669f7596e417a65 : SUCCESS in 7m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d687eb866d1d4531b4d3cb6d874f4232 : SUCCESS in 7m 40s","accounts_in_message":[],"_revision_number":1},{"id":"cfc5b6ff1c5e209161932adb7dc4bcf6c9c20367","author":{"_account_id":6732,"name":"Lingxian Kong","email":"anlin.kong@gmail.com","username":"kong"},"date":"2020-01-27 21:52:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"224cbbcb2c4fa2dbcb72093e91e8de481ced5ede","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-28 10:10:51.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nHello,\n\nFew questions/suggests inline","accounts_in_message":[],"_revision_number":1},{"id":"0ee5d644ea458483bbb4130a2998bce38dab0b52","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-28 10:17:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e9682350f540c35aba898a74cf0638e2bedf864b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-28 10:19:32.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a4ad4a5c4c2a96ba11ff063d3a83a1c8d117dec3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-28 10:22:01.000000000","message":"Patch Set 3: Workflow-1\n\n@Hervé: all good points, thanks. I updated my patch.\n\n@Ceilometer team: We might want some release notes though? W-1 on my patch in the meanwhile.","accounts_in_message":[],"_revision_number":3},{"id":"64d21d7ca7404b15acf1de288fa6d89e64829e97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-28 10:35:39.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/693ea273158146f3a8b2a3b6fa008076 : SUCCESS in 11m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/212134fe99704238981d3d6ddfa6e73a : SUCCESS in 5m 49s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/4944cbc30c3b4fdc8e9e8a3e35704ac9 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":3},{"id":"dada79bbc754ce4318376d58e6587afb438f919f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-01-29 08:29:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d39f46ad43bcd7743a84a64d51726881812a91e2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-29 13:21:47.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nHello,\n\nSee my inline comment","accounts_in_message":[],"_revision_number":3},{"id":"0ac253a0195adb1749a24bc0e98bccba464de808","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-04 18:20:17.000000000","message":"Patch Set 3:\n\nI submitted a release note upstream (to be backported to stable branches) so that the releases notes show some description of the changes: https://review.opendev.org/#/c/705790/","accounts_in_message":[],"_revision_number":3},{"id":"e63d0ee08ed788687dafef54c2701d3f736c117b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-12 13:29:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f5ec9b8ea7d52c90de831cd3a335da58134c6f40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-12 13:44:38.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7d117ee40c84b2585b957fc253fee9b : SUCCESS in 12m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/32132516f23e4e539fe54960c07e0b1f : FAILURE in 9m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/08eb1b1d6fdc47e599888f7bd80f5d18 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":4},{"id":"2af17c3dba925729effa62ae887d76fcd4453cf8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-12 15:49:20.000000000","message":"Patch Set 4:\n\nHum I think that the CI job failure is something similar to [1], I think we should ignore reno for stable branches like this one...\n\n[1] https://review.opendev.org/#/c/706626/\n\nElse thanks to addressed my comments :)","accounts_in_message":[],"_revision_number":4},{"id":"7833818602cd3f6b782b97aa1c6362edf433c100","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-12 16:33:10.000000000","message":"Patch Set 4:\n\n\u003e Hum I think that the CI job failure is something similar to [1], I\n \u003e think we should ignore reno for stable branches like this one...\n \u003e \n \u003e [1] https://review.opendev.org/#/c/706626/\n \u003e \n \u003e Else thanks to addressed my comments :)\n\nThanks for the pointer. Doug\u0027s fix work, I am proposing it to ceilometer.","accounts_in_message":[],"_revision_number":4},{"id":"bae7bb4c87c83984caf1f09a45aa6c29c7f1ce32","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-12 18:08:37.000000000","message":"Patch Set 4:\n\nPlease take a look to:\n- https://review.opendev.org/707443\n- https://review.opendev.org/707445\n- https://review.opendev.org/707444\n- https://review.opendev.org/707447\n\nTo fix the reno issue and so the CI job too.","accounts_in_message":[],"_revision_number":4},{"id":"812ddb4a66de87bf048c0667f59dc13693c0ab4a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-12 18:15:36.000000000","message":"Patch Set 4:\n\n@Hervé: I proposed similar patches earlier:\n\nhttps://review.opendev.org/#/c/707422/\nhttps://review.opendev.org/#/c/707423/\nhttps://review.opendev.org/#/c/707424/\nhttps://review.opendev.org/#/c/707425/","accounts_in_message":[],"_revision_number":4},{"id":"0bb74267398f013a407bad0431a6abcadfbb53d7","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-13 09:10:48.000000000","message":"Patch Set 4:\n\n\u003e @Hervé: I proposed similar patches earlier:\n \u003e \n \u003e https://review.opendev.org/#/c/707422/\n \u003e https://review.opendev.org/#/c/707423/\n \u003e https://review.opendev.org/#/c/707424/\n \u003e https://review.opendev.org/#/c/707425/\n\nYes thanks I missed them when I submitted mine.\n\nAnyway now I think you should point on these commits SHA in your different release versions, I don\u0027t think `recheck` will help us here but let\u0027s retry though.\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"77a36e7d928fb9f8055ee4236c6547bb52ff5f86","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-19 10:30:47.000000000","message":"Patch Set 4:\n\n\u003e \u003e @Hervé: I proposed similar patches earlier:\n \u003e \u003e\n \u003e \u003e https://review.opendev.org/#/c/707422/\n \u003e \u003e https://review.opendev.org/#/c/707423/\n \u003e \u003e https://review.opendev.org/#/c/707424/\n \u003e \u003e https://review.opendev.org/#/c/707425/\n \u003e \n \u003e Yes thanks I missed them when I submitted mine.\n \u003e \n \u003e Anyway now I think you should point on these commits SHA in your\n \u003e different release versions, I don\u0027t think `recheck` will help us\n \u003e here but let\u0027s retry though.\n \u003e \n \u003e recheck\n\nYeah, it would still build on the same sha. Need to update shas...","accounts_in_message":[],"_revision_number":4},{"id":"c21412de5f681cac3b575df23c3d22c07cd68e0e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-21 07:13:11.000000000","message":"Patch Set 4:\n\nThere are still some issues in ceilometer CI to fix before this can be updated.","accounts_in_message":[],"_revision_number":4},{"id":"2a6db7ed7014d06db7fa710d33aee11e08b0315e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-21 07:13:14.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"5fd51a43db6c1cc33ffddf01fdb08e63629d85dd","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-02-26 03:12:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"47caf76ceccdfa0d9a7c5993b997abe9929cb645","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 03:27:51.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca7cab5dbf4840aeb9f2169a23b5c16b : SUCCESS in 13m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4cb35512f7d64935944f2a7c5efe9f2a : FAILURE in 9m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7c324d45575f41aab18a35f5a9d4a0b9 : SUCCESS in 7m 34s","accounts_in_message":[],"_revision_number":4},{"id":"4081ebda1507cd8b38eb732978da082e65703211","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-02-26 08:29:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"19ae614e9d82a6f133e8440efbc2984d3588f0fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 08:45:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b0af1fac03d4efd858430faa7026da6 : SUCCESS in 11m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4f40313396d4eceaf303144fdc01690 : SUCCESS in 10m 26s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/5b51dac7c98f4821acd14b49cf7dfd05 : SUCCESS in 7m 06s","accounts_in_message":[],"_revision_number":5},{"id":"0aaf730db4b6ea875dffd64d51c949798108ddb1","author":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"date":"2020-02-26 09:16:07.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"1fce6ebe6ff5e4d30edc630fb7fc783af2a3b0f9","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 09:17:54.000000000","message":"Patch Set 5: PTL-Approved+1","accounts_in_message":[],"_revision_number":5},{"id":"23ee2e16c5f87a3e693522991743d3db1a9d3bad","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 21:22:28.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"e51fc1e545478db811aa70225ad002eece40d5ff","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-27 12:09:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"505174d9d9a4cf2c10a88fa18f2ab20e83bc3fb6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 12:09:40.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"648c3b10e114c70961d5b035e4a57d67dd4cc775","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 12:34:46.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f4f3b2de3a5465faa7a298b8a273aa9 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10ca61bc645749eea1bbe7cdb9e9394f : SUCCESS in 12m 59s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/85839d2de3fa42dbbc6fc3c51208f0df : SUCCESS in 13m 01s","accounts_in_message":[],"_revision_number":5},{"id":"0acfcd04d59f5622c2368d2feacf685e5944ca3a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 12:34:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"78d30994940951b53b627179a9b8a156ebcba03f","revisions":{"d0f55de25a7ece341b50b95bb9bb032bd0a448ec":{"kind":"REWORK","_number":1,"created":"2020-01-27 16:59:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/58/704358/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/704358/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/704358/1"}}},"commit":{"parents":[{"commit":"76404b0c948beea9bf0d05443ab0abd5f968084b","subject":"Merge \"Enable check-release-approval on experimental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76404b0c948beea9bf0d05443ab0abd5f968084b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-01-27 16:58:37.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-01-27 16:58:37.000000000","tz":60},"subject":"Release Ceilometer stable branches","message":"Release Ceilometer stable branches\n\nThis patch proposes releases for the following stable branches:\n\nRocky 11.0.2\nStein 12.1.0\nTrain 13.0.1\n\nWe bump the minor version for Stein as the branch includes a backport of\n`ceilometer-status upgrade check`.\n\nChange-Id: I6b128ff0d22ce13e1245a4c55ce608142564207d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0f55de25a7ece341b50b95bb9bb032bd0a448ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d0f55de25a7ece341b50b95bb9bb032bd0a448ec"}]},"branch":"refs/heads/master"},"0e2167c7860a04ce06799077ce68789a3e12d104":{"kind":"REWORK","_number":2,"created":"2020-01-28 10:17:58.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/58/704358/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/704358/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/704358/2"}}},"commit":{"parents":[{"commit":"76404b0c948beea9bf0d05443ab0abd5f968084b","subject":"Merge \"Enable check-release-approval on experimental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76404b0c948beea9bf0d05443ab0abd5f968084b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-01-28 10:17:38.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-01-28 10:17:38.000000000","tz":60},"subject":"Release Ceilometer stable branches","message":"Release Ceilometer stable branches\n\nThis patch proposes releases for the following stable branches:\n\nRocky 11.0.2\nStein 12.1.0\nTrain 13.0.1\n\nWe bump the minor version for Stein as the branch includes a backport of\n`ceilometer-status upgrade check`.\n\nChange-Id: I6b128ff0d22ce13e1245a4c55ce608142564207d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e2167c7860a04ce06799077ce68789a3e12d104"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0e2167c7860a04ce06799077ce68789a3e12d104"}]},"branch":"refs/heads/master"},"284e445ab8f366a89d6bccda82871c5d69111f94":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-28 10:19:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/58/704358/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/704358/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/704358/3"}}},"commit":{"parents":[{"commit":"76404b0c948beea9bf0d05443ab0abd5f968084b","subject":"Merge \"Enable check-release-approval on experimental\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/76404b0c948beea9bf0d05443ab0abd5f968084b"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-01-28 10:18:45.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-01-28 10:18:45.000000000","tz":60},"subject":"Release Ceilometer stable branches","message":"Release Ceilometer stable branches\n\nThis patch proposes releases for the following stable branches:\n\nRocky 11.1.0\nStein 12.1.0\nTrain 13.1.0\n\nWe bump the minor version for all branches since they backport some\nfeatures, such as adding the loadbalancer resource type.\n\nChange-Id: I6b128ff0d22ce13e1245a4c55ce608142564207d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/284e445ab8f366a89d6bccda82871c5d69111f94"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/284e445ab8f366a89d6bccda82871c5d69111f94"}]},"branch":"refs/heads/master"},"b83b8274ef4b6c65517fb72b6bb418e077b4050c":{"kind":"REWORK","_number":4,"created":"2020-02-12 13:29:59.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/58/704358/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/704358/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/704358/4"}}},"commit":{"parents":[{"commit":"dbd318033b579ab8baa2c33b84a17c5eaf5d2c8f","subject":"Merge \"[magnum] Ussuri milestone-2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/dbd318033b579ab8baa2c33b84a17c5eaf5d2c8f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-12 13:28:52.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-12 13:28:52.000000000","tz":60},"subject":"Release Ceilometer stable branches","message":"Release Ceilometer stable branches\n\nThis patch proposes releases for the following stable branches:\n\nRocky 11.1.0\nStein 12.1.0\nTrain 13.1.0\n\nWe bump the minor version for all branches since some commits declare\nnew resource types for Gnocchi, which could be interpreted as new\nfeatures.\n\nChange-Id: I6b128ff0d22ce13e1245a4c55ce608142564207d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b83b8274ef4b6c65517fb72b6bb418e077b4050c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b83b8274ef4b6c65517fb72b6bb418e077b4050c"}]},"branch":"refs/heads/master"},"78d30994940951b53b627179a9b8a156ebcba03f":{"kind":"REWORK","_number":5,"created":"2020-02-26 08:29:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/58/704358/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/58/704358/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/58/704358/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/58/704358/5"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-26 08:25:25.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-02-26 08:25:25.000000000","tz":60},"subject":"Release Ceilometer stable branches","message":"Release Ceilometer stable branches\n\nThis patch proposes releases for the following stable branches:\n\nRocky 11.1.0\nStein 12.1.0\nTrain 13.1.0\n\nWe bump the minor version for all branches since some commits declare\nnew resource types for Gnocchi, which could be interpreted as new\nfeatures.\n\nChange-Id: I6b128ff0d22ce13e1245a4c55ce608142564207d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78d30994940951b53b627179a9b8a156ebcba03f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/78d30994940951b53b627179a9b8a156ebcba03f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
