)]}'
{"id":"openstack%2Freleases~704988","triplet_id":"openstack%2Freleases~master~Ieab23754dfe45f449a2457f4d877efcb38bc7ae9","project":"openstack/releases","branch":"master","topic":"oslo_drop_py2_support","hashtags":[],"change_id":"Ieab23754dfe45f449a2457f4d877efcb38bc7ae9","subject":"Release oslo.log 4.0.0","status":"MERGED","created":"2020-01-30 12:31:55.000000000","updated":"2020-02-03 10:53:18.000000000","submitted":"2020-02-03 10:53:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"704988-1580727198101-71032318","meta_rev_id":"1722247065db096e4dd3ce112b82ce25b1fa1d9e","_number":704988,"virtual_id_number":704988,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-03 10:53:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-02-03 10:36:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2020-02-03 10:36:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-03 10:36:22.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-02-03 10:53:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d972f3d94809f09a4cd944c3a5ff20efa4060266","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-30 12:31:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2dc0f8e7869120b75913a13d6fcf3938c0d76bf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 12:46:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e2702415a647ebbef69d3e698d8985 : SUCCESS in 11m 31s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c4fd82c51a646a8917e06f2aeed506a : FAILURE in 7m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/151bb79c1ddb48339da87baa672acba9 : FAILURE in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"97de32ae1011a5571a64cf8090fed91dd6085338","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-30 17:20:22.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c4a5b6d7807097793596e1b9f5457caf289c37c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-30 17:34:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22d58f5902f540ad8bac42163cfac462 : SUCCESS in 11m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/62275d9c202b4ccca01821768b61a0c8 : FAILURE in 6m 24s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/58df9450626b4b869031d891cc0fdd48 : FAILURE in 6m 47s","accounts_in_message":[],"_revision_number":1},{"id":"7160fd079ed8a4163319802fe6e6a27ea671a564","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-01-31 09:39:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"331fc4dee70b12363908a031a1d26939d4013718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 09:54:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbb5dd14671b4eba973c97dd51cb1568 : SUCCESS in 9m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/28b1392224a34645a2c9d2c9f3e964a7 : FAILURE in 6m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f97ba81cf21340489afc94175e24b2ac : FAILURE in 7m 17s","accounts_in_message":[],"_revision_number":1},{"id":"c2b3b47c6c98f28e13864fb8db8292d4bffaa6a4","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-03 09:11:50.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1922693436ddd2807d79eab8fda54ac301c4897a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-03 09:16:19.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f661c9dfe5b4be64052eecce1fbf90b0c779be65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 09:30:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd9b646c21894ed5a43c0ba01c3ae54c : SUCCESS in 9m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/df3cdeb2c0394e95ba472656dbea77a3 : SUCCESS in 5m 23s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3582b9daa8c743f8841662a462b81416 : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":2},{"id":"7562a6be85cc97d48c4b80f9da4d016de01df26e","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-03 10:36:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99f3420560d519dee7bf8fde6811146439c9c741","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 10:36:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6f91bb22df677ac0ab5f4f4023f42f6198dbc7fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 10:53:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffcc586d9b454859b115d16ebd9e3561 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7b07e26bb4a4fbfbd270d6da51287c5 : SUCCESS in 10m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c47238ddbb854785b9f0e6f5baa15f41 : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":2},{"id":"804fb4c9880064e14c1b289c8a20fbc7ef3a4b67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 10:53:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ffb25a8371b727abf9585d3696885cddaee11c6c","revisions":{"ae4f79d919f05e63182ba9458eb4af74d468186d":{"kind":"REWORK","_number":1,"created":"2020-01-30 12:31:55.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/88/704988/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/704988/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/704988/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/704988/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/704988/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/704988/1"}}},"commit":{"parents":[{"commit":"090b0fee6188e5e9de64f32defbd7055ab5fa3eb","subject":"Merge \"Remove directly-released infra deliverables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/090b0fee6188e5e9de64f32defbd7055ab5fa3eb"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-30 12:30:06.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-30 12:30:06.000000000","tz":60},"subject":"Release oslo.log 4.0.0","message":"Release oslo.log 4.0.0\n\nThis version officially drop the python 2 support so we release a major\nversion of oslo.log to inform user that we break\nthe backward compatibility.\n\nChange-Id: Ieab23754dfe45f449a2457f4d877efcb38bc7ae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae4f79d919f05e63182ba9458eb4af74d468186d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ae4f79d919f05e63182ba9458eb4af74d468186d"}]},"branch":"refs/heads/master"},"ffb25a8371b727abf9585d3696885cddaee11c6c":{"kind":"REWORK","_number":2,"created":"2020-02-03 09:16:19.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/88/704988/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/704988/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/704988/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/704988/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/704988/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/704988/2"}}},"commit":{"parents":[{"commit":"090b0fee6188e5e9de64f32defbd7055ab5fa3eb","subject":"Merge \"Remove directly-released infra deliverables\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/090b0fee6188e5e9de64f32defbd7055ab5fa3eb"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-01-30 12:30:06.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-03 09:16:08.000000000","tz":60},"subject":"Release oslo.log 4.0.0","message":"Release oslo.log 4.0.0\n\nThis version officially drop the python 2 support so we release a major\nversion of oslo.log to inform user that we break\nthe backward compatibility.\n\nChange-Id: Ieab23754dfe45f449a2457f4d877efcb38bc7ae9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ffb25a8371b727abf9585d3696885cddaee11c6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ffb25a8371b727abf9585d3696885cddaee11c6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
