)]}'
{"id":"openstack%2Freleases~705260","triplet_id":"openstack%2Freleases~master~I29d6241e8619eac78c4e69ec617432a49f666cbf","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I29d6241e8619eac78c4e69ec617432a49f666cbf","subject":"Swift 2.24.0 release","status":"MERGED","created":"2020-01-31 17:30:13.000000000","updated":"2020-02-03 10:53:17.000000000","submitted":"2020-02-03 10:53:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705260-1580727197308-fa858c1e","meta_rev_id":"c61e6b633a2e65ac31c9c3d94aebc1d6ce675489","_number":705260,"virtual_id_number":705260,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-03 10:53:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-02-03 10:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2020-02-03 10:36:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-03 10:36:01.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-02-03 10:53:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f76f1322581ccd0e85d5ef2a81f0f5a0cebbd6d8","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2020-01-31 17:30:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22e2f33957a2104ea0d12c0cf848eb6a7fbbeca7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-31 17:43:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/722d26d8c8964216ab5f5fb63f830038 : SUCCESS in 12m 58s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/78822004ea784ad785fddff4ba33f94e : SUCCESS in 6m 35s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/67b65b0bbc414778a5544537affb69f6 : SUCCESS in 8m 12s","accounts_in_message":[],"_revision_number":1},{"id":"27869af10cff318cf35a1d7e281b2a431144f814","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-03 10:36:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"acb74d2f16222c20f6efaf8bafff5f9ddc294e80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 10:36:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c4fafa5925e5f1e7657f6f92c10138a890b8f74d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 10:53:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c61e6b633a2e65ac31c9c3d94aebc1d6ce675489","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 10:53:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aafa4032a29944ebab09b59e8e21cdb4 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69abf4ba6f5747ddb316e3ad36b6ba8c : SUCCESS in 10m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2ec28a2af74409daf2b61d31437544b : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"872a04a2bba803e3ccb4f0962660aa3a943d7062","revisions":{"872a04a2bba803e3ccb4f0962660aa3a943d7062":{"kind":"REWORK","_number":1,"created":"2020-01-31 17:30:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/60/705260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/60/705260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/705260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/705260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/60/705260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/60/705260/1"}}},"commit":{"parents":[{"commit":"631ae67077b32e96754b62f5d865f8462bdb71e6","subject":"Merge \"Bump tripleo-ansible to 1.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/631ae67077b32e96754b62f5d865f8462bdb71e6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-31 17:30:04.000000000","tz":-480},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2020-01-31 17:30:04.000000000","tz":-480},"subject":"Swift 2.24.0 release","message":"Swift 2.24.0 release\n\nChange-Id: I29d6241e8619eac78c4e69ec617432a49f666cbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/872a04a2bba803e3ccb4f0962660aa3a943d7062"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/872a04a2bba803e3ccb4f0962660aa3a943d7062"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
