)]}'
{"id":"openstack%2Freleases~705424","triplet_id":"openstack%2Freleases~master~I0eab7f690dd8f44f2c30df6fba5fca625406acff","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I0eab7f690dd8f44f2c30df6fba5fca625406acff","subject":"nova: Release 19.1.0","status":"MERGED","created":"2020-02-03 11:37:35.000000000","updated":"2020-02-06 12:38:51.000000000","submitted":"2020-02-06 12:38:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"705424-1580992731929-e685ee51","meta_rev_id":"1d05c4993b9925deac11c7b2dcdacb0137effdac","_number":705424,"virtual_id_number":705424,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-06 12:38:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-05 21:18:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-05 17:02:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":2,"date":"2020-02-06 12:09:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-02-03 14:45:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":1,"date":"2020-02-05 09:41:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":1,"date":"2020-02-06 12:09:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-02-03 11:38:41.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-03 11:38:41.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-02-03 14:45:02.000000000","updated_by":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"reviewer":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"state":"REVIEWER"},{"updated":"2020-02-05 09:41:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-05 17:02:13.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"REVIEWER"},{"updated":"2020-02-05 21:18:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-06 12:09:06.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-02-06 12:38:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"71bad5b8daba7b7bcba34e361af59ca64acb0997","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2020-02-03 11:37:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"723f313f81ffd4e353314d8f07c705ae40a6060b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 11:53:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3186f6b821774c9db38f9e523b533bc4 : SUCCESS in 9m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a1350bffadf749b2b62d208c0f0ac350 : SUCCESS in 10m 27s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/868ffcb586494b9b80fd54041a993925 : SUCCESS in 8m 09s","accounts_in_message":[],"_revision_number":1},{"id":"aa13003009ea5b50bd5499f5347ab8cda7ebb369","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-03 14:43:37.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nPTL ack","accounts_in_message":[],"_revision_number":1},{"id":"f2455ece044503bf60a9610b96be94f01a032cca","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2020-02-03 14:45:02.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"49b6571affa14af5deb9c93b6dfcdbe720f7f198","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-05 09:41:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5fc4109bc49d1d677c98ebe02194f8a2176bcb97","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2020-02-05 17:02:13.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"886350f5a138e72149da9adccc004a31bc922eb6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-05 21:18:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aab336d2c01e4d71bce8b2c73d925978bfc9b791","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-06 12:09:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"537f1d1e069178ed8e2fd223b9e2437d1e714399","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 12:09:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d99bc58649a5f1acdfad2bdcc17d567e6df3fd80","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 12:38:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db9ff09b190e45929e153afa5bcee54e : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79d3190ed97442ac94173d1da2f421c7 : SUCCESS in 9m 24s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5c41b37608024ca1a7ca0e2a8f6a38be : SUCCESS in 24m 22s","accounts_in_message":[],"_revision_number":1},{"id":"1d05c4993b9925deac11c7b2dcdacb0137effdac","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-06 12:38:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aa8c20c80335d63907834f1cd9f997bd87ec9feb","revisions":{"aa8c20c80335d63907834f1cd9f997bd87ec9feb":{"kind":"REWORK","_number":1,"created":"2020-02-03 11:37:35.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/24/705424/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/24/705424/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/705424/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/705424/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/24/705424/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/24/705424/1"}}},"commit":{"parents":[{"commit":"492059a48b613b1157057a6e8f43520b0c55b9ac","subject":"Merge \"Release oslo.log 4.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/492059a48b613b1157057a6e8f43520b0c55b9ac"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-02-03 11:34:27.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2020-02-03 11:37:16.000000000","tz":0},"subject":"nova: Release 19.1.0","message":"nova: Release 19.1.0\n\nMinor version bump due to the change in API behvaiour introduced by\nI0322d872bdff68936033a6f5a54e8296a6fb3434 / 4a691c33d1.\n\n$ git log --oneline --no-merges 19.0.3..HEAD\nd8fbf04f32 Mask the token used to allow access to consoles\n8346c527b3 FUP for in-place numa rebuild\n4a691c33d1 Disable NUMATopologyFilter on rebuild\n9f57d16f38 Block rebuild when NUMA topology changed\nd41afc7276 Remove \u0027test_cold_migrate_with_physnet_fails\u0027 test\n23d3b8fbc9 Remove global state from the FakeDriver\n6310e69303 libvirt: check job status for VIR_DOMAIN_EVENT_SUSPENDED_MIGRATED event\n2f48bbd224 libvirt: remove conditional on VIR_DOMAIN_EVENT_SUSPENDED_POSTCOPY\n99aa8ebc12 Zuul v3: use devstack-plugin-nfs-tempest-full\n95bf4a1e15 block_device: Copy original volume_type when missing for snapshot based volumes\n9f9f8d330a Do not update root_device_name during guest config\nfec95a2e4f Improve metadata server performance with large security groups\n99048af872 Cache security group driver\ne858ab3bbe compute: Use long_rpc_timeout in reserve_block_device_name\nad873b67b3 [stable-only] Disable cinder-backup service in nova-next job\n92eb45ba5e Replace time.sleep(10) with service forced_down in tests\n304d3f62a4 compute: Take an instance.uuid lock when rebooting\n603171bd5c Join migration_context and flavor in Migration.instance\na0290858b7 Block deleting compute services with in-progress migrations\n9983b24624 Add functional recreate revert resize test for bug 1852610\n7d67387246 Add functional recreate test for bug 1852610\nb18e42d20b Add functional recreate test for bug 1829479 and bug 1817833\n65ada9745c Use admin neutron client to see if instance has qos ports\nb1ae940ddb Use admin neutron client to query ports for binding\n87f7a4268f Add BFV wrinkle to TestNovaManagePlacementHealAllocations\n92645be4bb Add --instance option to heal_allocations\n4acec3e715 Add --dry-run option to heal_allocations CLI\ne00fa24a18 Don\u0027t delete compute node, when deleting service other than nova-compute\nf2d6dbd771 Revert \"openstack server create\" to \"nova boot\" in nova docs\na80eb82f67 Remove redundant call to get/create default security group\n844209b064 Add regression test for bug 1824435\n06caa62441 Add integration testing for heal_allocations\ndefe542eb1 Nova compute: add in log exception to help debug failures\n13d01b697d Imported Translations from Zanata\n8ee59604dc Avoid error 500 on shelve task_state race\n972d717984 Switch to opensuse-15 nodeset\n6272046bc8 Added openssh-client into bindep\n022ea28194 libvirt: Ignore volume exceptions during post_live_migration\n6038455e1d Fix listing deleted servers with a marker\n47caaccd4a Add functional regression test for bug 1849409\n792ec7c1df Fix unit of hw_rng:rate_period\nc0313bce06 Fix exception translation when creating volume\na06922d546 Stop sending bad values from libosinfo to libvirt\n530ad1ae88 Error out interrupted builds\ne4a5516098 Functional reproduce for bug 1833581\n080b4759d3 Prevent init_host test to interfere with other tests\nc565cab479 Add functional test for resize crash compute restart revert\n53bcf0b1ee Handle get_host_availability_zone error during reschedule\n58ad6aba98 Ignore sqla-migrate inspect.getargspec deprecation warnings on py36\n7e074ad030 rt: soften warning case in _remove_deleted_instances_allocations\n\nChange-Id: I0eab7f690dd8f44f2c30df6fba5fca625406acff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aa8c20c80335d63907834f1cd9f997bd87ec9feb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/aa8c20c80335d63907834f1cd9f997bd87ec9feb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
