)]}'
{"id":"openstack%2Freleases~706473","triplet_id":"openstack%2Freleases~master~I525a984a9406ea695bb815435ae7db4f8583f0af","project":"openstack/releases","branch":"master","topic":"gov-sync-cleanup","hashtags":[],"change_id":"I525a984a9406ea695bb815435ae7db4f8583f0af","subject":"Remove files for non-OpenStack deliverables","status":"MERGED","created":"2020-02-07 11:28:48.000000000","updated":"2020-02-10 14:58:03.000000000","submitted":"2020-02-10 14:58:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706473-1581346683988-e685ee51","meta_rev_id":"3059a366b3b59dd18cb86d4e432749f55895c67f","_number":706473,"virtual_id_number":706473,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 14:58:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-07 13:25:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-10 14:38:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 14:38:13.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-10 14:58:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9e6f1ea8d696cc411d889a5992cb74f9779de1e2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-07 11:28:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e76caa1de2080ab20ce8f1e02462494df546261","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 11:40:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa086e60bc1c40c3aa6b6390c63b06c3 : SUCCESS in 10m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5541e8da24f4cc5ad059b1a88daa6f0 : SUCCESS in 5m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c2188d209ff7418e8857a5db45fef29e : SUCCESS in 5m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0e6cb0c16f4ada9bce50279684cd48dd1aefac52","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-07 13:25:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b884819fed9bfc1cc0a8e580cce4e1a520441fd1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 14:38:13.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"286a770f22011282b82580b4fd79106f2755302d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:38:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67c85b333ed5b42accae41f09fc9cc64842fe565","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:58:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cb283b44d04421f90a67b085fd6620a : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ee3aa756b694defa905690882578ace : SUCCESS in 12m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/c47bb8f515954f6f903384c86d563488 : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":1},{"id":"c933123e32d8ebc1b5d161ec805ad45ef8042131","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:58:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4aa9bbae1a50fb91b59225b34c70fcf93d858fa7","revisions":{"4aa9bbae1a50fb91b59225b34c70fcf93d858fa7":{"kind":"REWORK","_number":1,"created":"2020-02-07 11:28:48.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/73/706473/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/706473/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/706473/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/706473/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/706473/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/706473/1"}}},"commit":{"parents":[{"commit":"6000ce3c67c3de7c8083332c1667989812f65b0c","subject":"Merge \"Release OpenStack-Ansible Rocky\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6000ce3c67c3de7c8083332c1667989812f65b0c"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-07 11:23:07.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-07 11:23:07.000000000","tz":60},"subject":"Remove files for non-OpenStack deliverables","message":"Remove files for non-OpenStack deliverables\n\nbandit and doc8 have migrated to GitHub and have releases there.\nListing older releases here is confusing at best, so let\u0027s remove\nthem, especially as arguably they were never part of the \"release\".\n\nsqlalchemy-migrate is not under OpenStack governance, so arguably\nshould never have been listed here. The ACL allows its core team\nto tag releases directly without going through openstack/releases.\n\nfairy-slipper has long been abandoned, and has arguably never been part\nof the \"release\", so let\u0027s remove it.\n\nChange-Id: I525a984a9406ea695bb815435ae7db4f8583f0af\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4aa9bbae1a50fb91b59225b34c70fcf93d858fa7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4aa9bbae1a50fb91b59225b34c70fcf93d858fa7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
