)]}'
{"id":"openstack%2Freleases~706501","triplet_id":"openstack%2Freleases~master~I979950b13f1dab75342e18d2ebae76d5fe6ef46f","project":"openstack/releases","branch":"master","topic":"adjutant","hashtags":[],"change_id":"I979950b13f1dab75342e18d2ebae76d5fe6ef46f","subject":"Add adjutant deliverables","status":"MERGED","created":"2020-02-07 13:34:59.000000000","updated":"2020-02-19 17:42:24.000000000","submitted":"2020-02-19 17:42:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706501-1582134144840-bbef5f2b","meta_rev_id":"71483223f9accaa306215ede0ed71f54c7ec66d9","_number":706501,"virtual_id_number":706501,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 17:42:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-19 17:22:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-19 03:09:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":2,"date":"2020-02-19 11:32:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-19 17:22:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-02-19 17:24:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-19 03:09:46.000000000","updated_by":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"reviewer":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"state":"REVIEWER"},{"updated":"2020-02-19 11:32:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-19 17:42:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f9b0b1270f852ec29625fc02537099e78ea93aa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-07 13:34:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"86cec0881606aff995166c0931605876f012fc02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-07 13:45:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df5742760e43421dbf41f28e6f591217 : SUCCESS in 9m 09s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5160ae5f0ff407594fedda236adc8c9 : SUCCESS in 6m 01s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/64dd83909ba34990875b0b885b1e7aa0 : SUCCESS in 5m 44s","accounts_in_message":[],"_revision_number":1},{"id":"6b597801f523953f1c49836e79f885f7ad4c0538","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-07 15:27:57.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM seems right settled.\n\nAdjutant\u0027projects already have some (old) existing releases [1][2][3]... but I don\u0027t think it\u0027s an issue in an ussuri context and in this context.\n\nElse team is ok [4]\n\n[1] https://github.com/openstack/adjutant/releases\n[2] https://github.com/openstack/python-adjutantclient/releases\n[3] https://github.com/openstack/adjutant-ui/releases\n[4] https://github.com/openstack/governance/blob/master/reference/projects.yaml#L15,L23","accounts_in_message":[],"_revision_number":1},{"id":"17822929a782440e6144abb251cf680450d066d3","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-02-11 01:44:22.000000000","message":"Patch Set 1: Code-Review-1\n\nHey! I\u0027d ideally like to move Adjutant to intermediary releases with cycle, because with the limited number of devs working on it, we\u0027re unlikely to meet consistent RC points.","accounts_in_message":[],"_revision_number":1},{"id":"c7256456dc569db762e4868befceb89abdba60f2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-11 09:28:36.000000000","message":"Patch Set 1: -Code-Review\n\nI remove -2 to highlight the Adrian\u0027s comment","accounts_in_message":[],"_revision_number":1},{"id":"2858f1d2bffc6f4fc5b1dbb0be0583162837dbbf","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-11 13:04:22.000000000","message":"Patch Set 1:\n\n\u003e Hey! I\u0027d ideally like to move Adjutant to intermediary releases\n \u003e with cycle, because with the limited number of devs working on it,\n \u003e we\u0027re unlikely to meet consistent RC points.\n\nThanks Adrian. Is the team ready to do an intermediary release? For cycle-with-intermediary, we want to have at least one release by milestone 2 (this Thursday) to be included in the overall coordinated release. It can just be a 0.1.0. Just so we can make sure the release machinery is good and things are on track for a final release near the end of the cycle.","accounts_in_message":[],"_revision_number":1},{"id":"254c780c0d2fbb027f85052e9a50b7f112b97b87","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-02-13 02:10:19.000000000","message":"Patch Set 1:\n\n\u003e \u003e Hey! I\u0027d ideally like to move Adjutant to intermediary releases\n \u003e \u003e with cycle, because with the limited number of devs working on\n \u003e it,\n \u003e \u003e we\u0027re unlikely to meet consistent RC points.\n \u003e \n \u003e Thanks Adrian. Is the team ready to do an intermediary release? For\n \u003e cycle-with-intermediary, we want to have at least one release by\n \u003e milestone 2 (this Thursday) to be included in the overall\n \u003e coordinated release. It can just be a 0.1.0. Just so we can make\n \u003e sure the release machinery is good and things are on track for a\n \u003e final release near the end of the cycle.\n\nI have just one patch I feel I really need to merge before any release, but once that is done I can do one.\n\nJust doing some additional testing on:\nhttps://review.opendev.org/#/c/706723/\n\nAnd when that is merged I\u0027d love to do a release of Adjutant itself, and Adjutant-UI not too long after.","accounts_in_message":[],"_revision_number":1},{"id":"56548eecb3451f4a34de8b0d0bfadf8080bcedc5","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-02-19 03:09:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b45414e8b54e8031c09ff3756e9867ada9e2bf49","author":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"date":"2020-02-19 03:09:46.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"44da0311ff2beba451235896b12668e077e69bd4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 03:21:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3866aeab9a641dd9739b3e3a66a0fe9 : SUCCESS in 11m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4c382e7a1dc141fb9ed7e051549e68d9 : SUCCESS in 5m 25s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2608b78d61ec4856a39c8a32ffac69b7 : SUCCESS in 5m 45s","accounts_in_message":[],"_revision_number":2},{"id":"49b372f4b8ebe823b1efaf3ac7c67d41c37c75cf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-19 11:32:47.000000000","message":"Patch Set 2: Code-Review+2\n\nAdrian isn\u0027t seen as the PTL [1] I suppose it\u0027s due to the email address   used by Adrian [2] which differ from address used with LP/gerrit.\n\n[1] https://storage.gra.cloud.ovh.net/v1/AUTH_dcaab5e32b234d56b626f72581e3644c/zuul_opendev_logs_260/706501/2/check/releases-tox-list-changes/2608b78/job-output.txt\n[2] https://github.com/openstack/governance/blob/master/reference/projects.yaml#L1,L5","accounts_in_message":[],"_revision_number":2},{"id":"77bf57680c5b2e89a89aaf6bf349ab3894ea9f27","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-19 11:44:18.000000000","message":"Patch Set 2:\n\nPTL approval will be fixed by https://review.opendev.org/#/c/708623/ (cf. my previous comment)","accounts_in_message":[],"_revision_number":2},{"id":"52ab978b3206b0747af1afed3b2f9b922e85a74c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-19 17:22:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nThanks Adrian. Please try to get an actual release proposed for these as soon as possible.\n\nhttps://releases.openstack.org/ussuri/schedule.html#u-mf","accounts_in_message":[],"_revision_number":2},{"id":"96004da118b3c4deaf9d738582ebea500f8ad734","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 17:22:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"77387642f328daa5a136b6639e77b73d9ae4c851","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 17:24:11.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"42f23046bf24b1900c50917688259c2643c24b96","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 17:42:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cc0edbce9e170d50377ebc4f6e6d92e64628a642","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 17:42:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07364485e98f4aa1a3bece3df3005f87 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3537bebb4e804e3cb467880b71ddd042 : SUCCESS in 13m 47s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed0c742257594d5aae72d7c20bbf5f31 : SUCCESS in 8m 36s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d97113a1b8ada83e6215c2f5a117dc4bcae81b6e","revisions":{"6cd9b0d4d405f8b1aab6db10819ee41952c0379d":{"kind":"REWORK","_number":1,"created":"2020-02-07 13:34:59.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/01/706501/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/706501/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/706501/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/706501/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/706501/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/706501/1"}}},"commit":{"parents":[{"commit":"d9e126cbf49e70008afd02c1a972cccd672bdfba","subject":"Merge \"Release catellan 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9e126cbf49e70008afd02c1a972cccd672bdfba"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-07 13:34:32.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-07 13:34:32.000000000","tz":-360},"subject":"Add adjutant deliverables","message":"Add adjutant deliverables\n\nChange-Id: I979950b13f1dab75342e18d2ebae76d5fe6ef46f\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6cd9b0d4d405f8b1aab6db10819ee41952c0379d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6cd9b0d4d405f8b1aab6db10819ee41952c0379d"}]},"branch":"refs/heads/master"},"d97113a1b8ada83e6215c2f5a117dc4bcae81b6e":{"kind":"REWORK","_number":2,"created":"2020-02-19 03:09:31.000000000","uploader":{"_account_id":10420,"name":"Adrian Turjak","email":"devs+openstack@uncaught-exceptions.com","username":"adriant"},"ref":"refs/changes/01/706501/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/01/706501/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/706501/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/706501/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/01/706501/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/01/706501/2"}}},"commit":{"parents":[{"commit":"d9e126cbf49e70008afd02c1a972cccd672bdfba","subject":"Merge \"Release catellan 2.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9e126cbf49e70008afd02c1a972cccd672bdfba"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-07 13:34:32.000000000","tz":-360},"committer":{"name":"Adrian Turjak","email":"adriant@catalystcloud.nz","date":"2020-02-19 03:09:26.000000000","tz":780},"subject":"Add adjutant deliverables","message":"Add adjutant deliverables\n\nChange-Id: I979950b13f1dab75342e18d2ebae76d5fe6ef46f\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d97113a1b8ada83e6215c2f5a117dc4bcae81b6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d97113a1b8ada83e6215c2f5a117dc4bcae81b6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
