)]}'
{"id":"openstack%2Freleases~706769","triplet_id":"openstack%2Freleases~master~Ifd17c798010606da421bc4127f6b766c5c0c880e","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"Ifd17c798010606da421bc4127f6b766c5c0c880e","subject":"Release python-ironicclient 4.0.0 for Ussuri","status":"MERGED","created":"2020-02-10 09:52:35.000000000","updated":"2020-02-10 14:52:51.000000000","submitted":"2020-02-10 14:52:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706769-1581346371137-b4850ae3","meta_rev_id":"a3cea2129132404d823f81fd62659bd2fc1eeabe","_number":706769,"virtual_id_number":706769,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-02-10 13:08:55.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 14:52:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-10 14:34:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-02-10 13:46:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-10 14:34:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-02-10 13:09:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 13:46:09.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-10 14:34:38.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-10 14:52:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c381abb362c1b4532776f7c43e1f9e1dc8dfeb8f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-10 09:52:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85bd05f1006ec5a567aea43d02e3f9a0642c147b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:06:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43334349c1f24a36844b0026835b922f : SUCCESS in 9m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9ceba2c6831f46d7a50685c9b2dae95a : SUCCESS in 7m 17s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f1f681314b844c1d84129a92ffda731e : SUCCESS in 8m 40s","accounts_in_message":[],"_revision_number":1},{"id":"1a8a46df51cb42c8f26a3d8b63cd2aa96eb71fb2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-10 10:09:19.000000000","message":"Patch Set 1: Workflow-1\n\nWaiting for https://review.opendev.org/706776","accounts_in_message":[],"_revision_number":1},{"id":"90241e00d4ccb0a1f681b9cec7649fcedc8fe4ad","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:11:12.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"9f5c0fee9df652896179ecddd6fd212bcdbc15a9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-10 13:08:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"debdb100f55604ececd2026b06ff8efac501b52f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-10 13:08:55.000000000","message":"Patch Set 2:\n\n\u003e Waiting for https://review.opendev.org/706776\n\nLanded","accounts_in_message":[],"_revision_number":2},{"id":"1a9d3d82a04d6c2fd08c514794c083541634b8dd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:09:57.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"b30d68ba8c8d7df600ada507f7a47770bdb7bce3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 13:21:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93d1337d990c4b9a85bba9e18efb8609 : SUCCESS in 9m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b10db87a9b474af4bb1cea1f7e866b89 : SUCCESS in 6m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9cd94619df69458abad7b3dfc8a48b1b : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":2},{"id":"654b25c928182f6a8b58f34eb26888e5e70a2e07","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-10 13:46:09.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM\n\nJust a side note Py2 was dropped with f51566545e848f4b09d066e56da04cafe28ccc2b and this is a major release so it look fine!","accounts_in_message":[],"_revision_number":2},{"id":"d540bd2ddf192f2363b73dd709f845547ec4eff1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 14:34:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"c5b202fcda05df96fe4d9fbfae84d6aac81758f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:34:57.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c914d0e1691c718730f12dd42fbf9a9a343e0d81","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:52:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b4f2b0c42e14c369dba3da372d3a575 : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7623d92987e4d39b50a8f2ce874e180 : SUCCESS in 9m 46s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/732c71e9404546d686b0a1dc8d6d2789 : SUCCESS in 8m 24s","accounts_in_message":[],"_revision_number":2},{"id":"a3cea2129132404d823f81fd62659bd2fc1eeabe","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:52:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"7102ae291107b06d623c452338b42ed79cb01a8a","revisions":{"28dd9028cf3635f695d62fd10c465fe57fb47eb7":{"kind":"REWORK","_number":1,"created":"2020-02-10 09:52:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/69/706769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/706769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/706769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/706769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/706769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/706769/1"}}},"commit":{"parents":[{"commit":"7e77af90f07e04adeb2dff42c93449fb23963097","subject":"Merge \"Run release-approval job on specific pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e77af90f07e04adeb2dff42c93449fb23963097"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-10 09:52:23.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-10 09:52:23.000000000","tz":0},"subject":"Release python-ironicclient 4.0.0 for Ussuri","message":"Release python-ironicclient 4.0.0 for Ussuri\n\nChange-Id: Ifd17c798010606da421bc4127f6b766c5c0c880e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28dd9028cf3635f695d62fd10c465fe57fb47eb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/28dd9028cf3635f695d62fd10c465fe57fb47eb7"}]},"branch":"refs/heads/master"},"7102ae291107b06d623c452338b42ed79cb01a8a":{"kind":"REWORK","_number":2,"created":"2020-02-10 13:08:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/69/706769/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/69/706769/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/706769/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/706769/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/69/706769/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/69/706769/2"}}},"commit":{"parents":[{"commit":"7e77af90f07e04adeb2dff42c93449fb23963097","subject":"Merge \"Run release-approval job on specific pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e77af90f07e04adeb2dff42c93449fb23963097"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-10 09:52:23.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-10 13:08:43.000000000","tz":0},"subject":"Release python-ironicclient 4.0.0 for Ussuri","message":"Release python-ironicclient 4.0.0 for Ussuri\n\nChange-Id: Ifd17c798010606da421bc4127f6b766c5c0c880e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7102ae291107b06d623c452338b42ed79cb01a8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7102ae291107b06d623c452338b42ed79cb01a8a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
