)]}'
{"id":"openstack%2Freleases~706773","triplet_id":"openstack%2Freleases~master~I7914e26bfd6d83056c6ab1950c533a38f54a9a32","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I7914e26bfd6d83056c6ab1950c533a38f54a9a32","subject":"Release ironic-python-agent 6.0.0 for Ussuri","status":"MERGED","created":"2020-02-10 09:58:29.000000000","updated":"2020-02-10 15:18:01.000000000","submitted":"2020-02-10 15:18:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"706773-1581347881381-2176cf01","meta_rev_id":"ca3bee0672a1eefce1d8b136df1d0fccae567c46","_number":706773,"virtual_id_number":706773,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-10 15:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-10 14:59:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-10 14:59:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-02-10 14:38:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-10 13:59:40.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-10 14:59:15.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-10 15:18:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f8932afabb5c49cb35198fdefb515eccc0d7d11f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-10 09:58:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d9d2e427017366ae607ef909cb3b2f5ac13b7a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 10:11:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e10b75b4add44190a633a3be8cd2cd17 : SUCCESS in 9m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b5776fecd870434893fe9fc9b3f41327 : SUCCESS in 7m 52s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/787cc42b7a7341f98a6d48002193b280 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":1},{"id":"b3531b503dea7086a68b3b8b013b31db86e132cd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-10 11:09:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"8bf5fffeee675198610d95698d517713df87d144","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-02-10 11:09:20.000000000","message":"Patch Set 2:\n\nUpdated to include release note tidy up.","accounts_in_message":[],"_revision_number":2},{"id":"9054169ccd6facc27e6cf5000d557f2307a7c330","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 11:10:54.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"3281691ec827814d02231bc92720dacd5605ed48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 11:25:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad8c7e68a2e74997a42d39e41bbd387b : SUCCESS in 13m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/780c769cc5174f93835512ffafff9cf0 : SUCCESS in 7m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1c9d6dc74d874a16a19e088652329b70 : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":2},{"id":"4925eb673e1da8d90692858dfbd9567f7e06f89a","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-10 13:59:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d4cec511d2eba87762ba485f9fc1baef2801a9cc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 14:36:14.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"41ea164586c11c8c5f23c42191b32c61f9ac634b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:38:34.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"854065b01ea00e68669cfa483cc88463c2e57df2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:56:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d23eac93e15346a8b46fab84ba22931e : SUCCESS in 12m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8e1155d29b924df3b6e3fad2ccc13c44 : SUCCESS in 6m 09s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/010cc81c1b7f4b319d9a594e9eb9b8ce : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":3},{"id":"cf79274f8e4b9f0007c2896eee729a789b4f75e0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 14:59:15.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8e99cdea27d6838e304af8159af6f2f15fc530ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 14:59:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"fa60e1a5c09ec459f55d56bfce971b95e95bb9a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 15:18:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94cc9304c9cb4298aa3cd10e18302511 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44e7b78a646a4bb694e1c640c606d43c : SUCCESS in 9m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2f1b00b75b4f4edeb7ca51d92f312048 : SUCCESS in 6m 52s","accounts_in_message":[],"_revision_number":3},{"id":"ca3bee0672a1eefce1d8b136df1d0fccae567c46","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-10 15:18:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7c5626e949bcc2318a2c5c200594f06a06f1f3fa","revisions":{"6a97bffdb9d6adf87e24159b93612befde250c54":{"kind":"REWORK","_number":1,"created":"2020-02-10 09:58:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/706773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/706773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/706773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/706773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/706773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/706773/1"}}},"commit":{"parents":[{"commit":"7e77af90f07e04adeb2dff42c93449fb23963097","subject":"Merge \"Run release-approval job on specific pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e77af90f07e04adeb2dff42c93449fb23963097"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-10 09:58:11.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-10 09:58:11.000000000","tz":0},"subject":"Release ironic-python-agent 14.0.0 for Ussuri","message":"Release ironic-python-agent 14.0.0 for Ussuri\n\nChange-Id: I7914e26bfd6d83056c6ab1950c533a38f54a9a32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a97bffdb9d6adf87e24159b93612befde250c54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6a97bffdb9d6adf87e24159b93612befde250c54"}]},"branch":"refs/heads/master"},"5c3cd72606ea35d14d5f31ff980a5e99e96a005f":{"kind":"REWORK","_number":2,"created":"2020-02-10 11:09:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/73/706773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/706773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/706773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/706773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/706773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/706773/2"}}},"commit":{"parents":[{"commit":"7e77af90f07e04adeb2dff42c93449fb23963097","subject":"Merge \"Run release-approval job on specific pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e77af90f07e04adeb2dff42c93449fb23963097"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-10 09:58:11.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-10 11:09:02.000000000","tz":0},"subject":"Release ironic-python-agent 14.0.0 for Ussuri","message":"Release ironic-python-agent 14.0.0 for Ussuri\n\nChange-Id: I7914e26bfd6d83056c6ab1950c533a38f54a9a32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5c3cd72606ea35d14d5f31ff980a5e99e96a005f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/5c3cd72606ea35d14d5f31ff980a5e99e96a005f"}]},"branch":"refs/heads/master"},"7c5626e949bcc2318a2c5c200594f06a06f1f3fa":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-02-10 14:36:14.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/73/706773/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/73/706773/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/706773/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/706773/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/73/706773/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/73/706773/3"}}},"commit":{"parents":[{"commit":"7e77af90f07e04adeb2dff42c93449fb23963097","subject":"Merge \"Run release-approval job on specific pipeline\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e77af90f07e04adeb2dff42c93449fb23963097"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-02-10 09:58:11.000000000","tz":0},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-10 14:36:13.000000000","tz":0},"subject":"Release ironic-python-agent 6.0.0 for Ussuri","message":"Release ironic-python-agent 6.0.0 for Ussuri\n\nChange-Id: I7914e26bfd6d83056c6ab1950c533a38f54a9a32\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7c5626e949bcc2318a2c5c200594f06a06f1f3fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7c5626e949bcc2318a2c5c200594f06a06f1f3fa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
