)]}'
{"id":"openstack%2Freleases~707003","triplet_id":"openstack%2Freleases~master~Id1f01506e80f2a3b07e0982e1b74f6ee73fa229c","project":"openstack/releases","branch":"master","topic":"ussuri-2","hashtags":[],"change_id":"Id1f01506e80f2a3b07e0982e1b74f6ee73fa229c","subject":"[congress] Ussuri milestone-2 release","status":"MERGED","created":"2020-02-10 23:45:02.000000000","updated":"2020-02-14 22:42:29.000000000","submitted":"2020-02-14 22:42:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707003-1581720149418-25e6796c","meta_rev_id":"8b290041980a3e1ce8c761525b8b2d1bb6e9a55c","_number":707003,"virtual_id_number":707003,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-14 22:42:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-14 22:28:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-11 10:41:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-14 22:28:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-02-11 13:18:26.000000000","_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 10:41:46.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-11 13:18:26.000000000","updated_by":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"reviewer":{"_account_id":18591,"name":"Eric Kao","email":"ekcs.openstack@gmail.com","username":"ekcs"},"state":"REVIEWER"},{"updated":"2020-02-14 22:42:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2687c7379f1aea395239836fe8110241a4b54bc","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 23:45:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bc32641a952550b4b09324e4d19b59597fe8efa8","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 23:47:27.000000000","message":"Topic changed from congress-ussuri-2 to ussuri-2","accounts_in_message":[],"_revision_number":1},{"id":"b15684d082ea408779bd2af65b308050a1db34e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 00:03:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/525e6e777a3d4343b503f009f9807ff2 : SUCCESS in 11m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40c4703633b245ed9e6262ccb1d1cd8b : SUCCESS in 7m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/18611204a57046739a291065739a2399 : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":1},{"id":"9a1de9c2ea36bf84bb7a0b42b192ad0b6da0d1af","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-11 10:41:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8f474a46ab87cc2186684af58934db06cc573ef6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-14 22:28:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nPast deadline and no new commits to repo since this was generated.","accounts_in_message":[],"_revision_number":1},{"id":"535f96b88b55b797e944b6fc84d79eec722475d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 22:28:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a6771fdb9d7e347524a4711dcecef133cd2968c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 22:42:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c48ad0ce425c4778938e27d9a6ecc9b9 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df89cda6c75944ccb197ca80563c9f69 : SUCCESS in 12m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e660eb8fa3f74759a1f0c7e57fdf0154 : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1},{"id":"440de8a16ae67c27dbba1e8709763140abffcb20","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 22:42:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50cfe728d162fd973a448e8bb75482c831c77435","revisions":{"50cfe728d162fd973a448e8bb75482c831c77435":{"kind":"REWORK","_number":1,"created":"2020-02-10 23:45:02.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/03/707003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/03/707003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/707003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/707003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/03/707003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/03/707003/1"}}},"commit":{"parents":[{"commit":"f9595d89c6babdc93c4448a1298539d560bff3f9","subject":"Merge \"\u0027Independent\u0027 PowerVM items no longer released\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9595d89c6babdc93c4448a1298539d560bff3f9"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-10 23:36:57.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-10 23:36:57.000000000","tz":-360},"subject":"[congress] Ussuri milestone-2 release","message":"[congress] Ussuri milestone-2 release\n\nThis adds a release to be included in Ussuri\n\nChange-Id: Id1f01506e80f2a3b07e0982e1b74f6ee73fa229c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/50cfe728d162fd973a448e8bb75482c831c77435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/50cfe728d162fd973a448e8bb75482c831c77435"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
