)]}'
{"id":"openstack%2Freleases~707007","triplet_id":"openstack%2Freleases~master~Ia04e5b58aba1eb5e9a7d7df1e769dba084eed33c","project":"openstack/releases","branch":"master","topic":"ussuri-2","hashtags":[],"change_id":"Ia04e5b58aba1eb5e9a7d7df1e769dba084eed33c","subject":"[glance] Ussuri milestone-2 release","status":"MERGED","created":"2020-02-10 23:45:09.000000000","updated":"2020-03-02 09:51:41.000000000","submitted":"2020-03-02 09:51:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707007-1583142701630-b16e9fe7","meta_rev_id":"ebc0fa814297daa915da22a1f962406dc4774a20","_number":707007,"virtual_id_number":707007,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-02 09:51:41.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-01 23:14:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2020-03-02 09:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2020-03-02 09:33:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 13:22:15.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-02-11 18:10:06.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-03-02 09:33:29.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-02 09:51:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba8a476a0883d8d56a8cb619442973e7fdb083ad","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 23:45:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1e78eaa1e52a99d37458e038bd46f3db6f805f1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 23:47:16.000000000","message":"Topic changed from glance-ussuri-2 to ussuri-2","accounts_in_message":[],"_revision_number":1},{"id":"52dc77b675a523b515818b43cb14c012b148f3a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 00:12:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a6210945aba41ee911a34b517eeeeb3 : SUCCESS in 10m 50s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a77f83cc48aa4ca9b300125f45b15e60 : SUCCESS in 8m 00s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f03d9ffe95034d47943063102f5ed50e : SUCCESS in 10m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5dc070ab1c8348dcce1d5e41d13380fc8dc9a8a2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-11 10:23:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ed80b806619516bc5e715b08b4e1c9ecb496badf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-11 18:10:06.000000000","message":"Patch Set 1: Code-Review-1\n\nWe are waiting for 1/2 patches to be reviewed and merged before release.","accounts_in_message":[],"_revision_number":1},{"id":"ba07bec4d9584272a7adff9476678f7ae9e5cf8c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-14 22:22:32.000000000","message":"Patch Set 1: Workflow-1\n\nWaiting for update from team.","accounts_in_message":[],"_revision_number":1},{"id":"53234ad493c6854aa0323b7920509e1dab8cd756","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-17 06:31:06.000000000","message":"Patch Set 1:\n\nNeed couple of days/or this week to get some client patches in before release.","accounts_in_message":[],"_revision_number":1},{"id":"ce8f990277324dbb51b3a80ce16505f8861359ee","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-19 17:18:25.000000000","message":"Patch Set 1:\n\nAny update on this Abhishek? Would be good to get a release out with the latest.","accounts_in_message":[],"_revision_number":1},{"id":"500d67de770aebd3a386e95e936a4a141100b00e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-19 18:36:46.000000000","message":"Patch Set 1:\n\nHi Sean,\n\nSorry for the delay, we need to wait for this week to get 2 important patches merged before we release.","accounts_in_message":[],"_revision_number":1},{"id":"c5d076e772fe0f94e82b02a2502b9849a4bec65d","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-20 16:13:57.000000000","message":"Patch Set 1: -Code-Review\n\nI dropping my +2 by waiting the 2 patches.","accounts_in_message":[],"_revision_number":1},{"id":"834d7da6a9f4187418fdb61a04334160eb9d0344","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-02-28 05:38:45.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"6236e7c7bd245cc49a43fe347f8906eac860a4bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 05:50:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39703092d7b34ec7bd102487a5b8c931 : SUCCESS in 11m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/01f958cf0f9f4a4983ef7bc4f61e9c3f : SUCCESS in 6m 15s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/7609fc2273fc4084a7cb5b589cd8ae15 : SUCCESS in 6m 25s","accounts_in_message":[],"_revision_number":2},{"id":"aa5e9849a02d53ab4f418ce700e22bca04daf977","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-01 23:14:09.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"fca23349af6415c1c2ab41a1a96c1856fba566a2","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-02 09:33:29.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cbafb226c752f6c6bb1254d4a4a56b835f34b3bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 09:33:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"08147b83e4cec53e51423abb2a199e7754913c08","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 09:51:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ebc0fa814297daa915da22a1f962406dc4774a20","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 09:51:41.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/045ed912503d4a5db9001938883b3fab : SUCCESS in 7m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8685ccd44734fe5a2874a86e5a5f3d7 : SUCCESS in 12m 10s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1821a1c868c244f8aba2cdea56b30d7c : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4f80972fb110ea70fe033a9d8634fc72ac9f7cae","revisions":{"63223448924495aa583031d45bda15ca1f20f729":{"kind":"REWORK","_number":1,"created":"2020-02-10 23:45:09.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/07/707007/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/707007/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/707007/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/707007/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/707007/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/707007/1"}}},"commit":{"parents":[{"commit":"f9595d89c6babdc93c4448a1298539d560bff3f9","subject":"Merge \"\u0027Independent\u0027 PowerVM items no longer released\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9595d89c6babdc93c4448a1298539d560bff3f9"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-10 23:36:58.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-10 23:36:58.000000000","tz":-360},"subject":"[glance] Ussuri milestone-2 release","message":"[glance] Ussuri milestone-2 release\n\nThis adds a release to be included in Ussuri\n\nChange-Id: Ia04e5b58aba1eb5e9a7d7df1e769dba084eed33c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63223448924495aa583031d45bda15ca1f20f729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/63223448924495aa583031d45bda15ca1f20f729"}]},"branch":"refs/heads/master"},"4f80972fb110ea70fe033a9d8634fc72ac9f7cae":{"kind":"REWORK","_number":2,"created":"2020-02-28 05:38:45.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/07/707007/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/07/707007/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/707007/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/707007/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/07/707007/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/07/707007/2"}}},"commit":{"parents":[{"commit":"f9595d89c6babdc93c4448a1298539d560bff3f9","subject":"Merge \"\u0027Independent\u0027 PowerVM items no longer released\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9595d89c6babdc93c4448a1298539d560bff3f9"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-10 23:36:58.000000000","tz":-360},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2020-02-28 05:37:43.000000000","tz":0},"subject":"[glance] Ussuri milestone-2 release","message":"[glance] Ussuri milestone-2 release\n\nThis adds a release to be included in Ussuri\n\nChange-Id: Ia04e5b58aba1eb5e9a7d7df1e769dba084eed33c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f80972fb110ea70fe033a9d8634fc72ac9f7cae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f80972fb110ea70fe033a9d8634fc72ac9f7cae"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
