)]}'
{"id":"openstack%2Freleases~707022","triplet_id":"openstack%2Freleases~master~I13ad2cfeef76ea145e49f9fd054a4dda9a11da68","project":"openstack/releases","branch":"master","topic":"ussuri-2","hashtags":[],"change_id":"I13ad2cfeef76ea145e49f9fd054a4dda9a11da68","subject":"[searchlight] Ussuri milestone-2 release","status":"MERGED","created":"2020-02-10 23:45:37.000000000","updated":"2020-02-11 13:42:00.000000000","submitted":"2020-02-11 13:42:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707022-1581428520495-e503d632","meta_rev_id":"6e0f8d2115dfe195d6c609d3efed5e1365ff9795","_number":707022,"virtual_id_number":707022,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-11 13:42:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-11 13:31:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":2,"date":"2020-02-11 10:03:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":1,"date":"2020-02-11 13:22:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-11 13:31:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-02-11 13:25:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-11 10:03:00.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-11 13:22:52.000000000","updated_by":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"reviewer":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"state":"REVIEWER"},{"updated":"2020-02-11 13:42:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"636109d992704a2ec6ee1e326566c56c5928f7aa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 23:45:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3304f646aa5f241c9520ca7ce6c34798c8ed544d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-10 23:48:51.000000000","message":"Topic changed from searchlight-ussuri-2 to ussuri-2","accounts_in_message":[],"_revision_number":1},{"id":"2882764cac7b7088ebf7e9847efa16823a91d36f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 00:22:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ddfcd1f11b124abd9fb395c20f5e8562 : SUCCESS in 10m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/7f564d94d9ac45a1b93e3eb48ebcbe5b : SUCCESS in 5m 50s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/1668011ed8c14f3891d9da13dedf386f : SUCCESS in 4m 37s","accounts_in_message":[],"_revision_number":1},{"id":"1a4e4cd34dd33d878dad4db9f084012fcf3e9042","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-11 10:03:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e86dff640fbae0f5a1f3737a4961f7d93f8ad60d","author":{"_account_id":27068,"name":"Trinh Nguyen","email":"dangtrinhnt@gmail.com","username":"dangtrinhnt"},"date":"2020-02-11 13:22:52.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM. Thanks.","accounts_in_message":[],"_revision_number":1},{"id":"23c9a055ba106632fb2a319f46b2643c590f6c39","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 13:25:00.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"9ced3a53d24d202586ab094c1e050425440cbf37","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-11 13:31:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":1},{"id":"6f4cefe4e715d92dfe14fb9b6241045c6811904d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 13:31:37.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c42a6c07390535526edce9122160bb3eef6b7dfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 13:42:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ec50ae9335e47a3aac05a0b144cbc8a : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d463c0be5d46452a9d623a3f8d92d988 : SUCCESS in 8m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9920f98185a34813a1211826410591b3 : SUCCESS in 5m 39s","accounts_in_message":[],"_revision_number":1},{"id":"6e0f8d2115dfe195d6c609d3efed5e1365ff9795","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-11 13:42:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cd607e6cabd20ee0c385c42f1e61c558e08ee0c","revisions":{"2cd607e6cabd20ee0c385c42f1e61c558e08ee0c":{"kind":"REWORK","_number":1,"created":"2020-02-10 23:45:37.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/22/707022/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/22/707022/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/707022/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/707022/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/22/707022/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/22/707022/1"}}},"commit":{"parents":[{"commit":"f9595d89c6babdc93c4448a1298539d560bff3f9","subject":"Merge \"\u0027Independent\u0027 PowerVM items no longer released\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/f9595d89c6babdc93c4448a1298539d560bff3f9"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-10 23:36:58.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-10 23:36:58.000000000","tz":-360},"subject":"[searchlight] Ussuri milestone-2 release","message":"[searchlight] Ussuri milestone-2 release\n\nThis adds a release to be included in Ussuri\n\nChange-Id: I13ad2cfeef76ea145e49f9fd054a4dda9a11da68\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2cd607e6cabd20ee0c385c42f1e61c558e08ee0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2cd607e6cabd20ee0c385c42f1e61c558e08ee0c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
