)]}'
{"id":"openstack%2Freleases~707605","triplet_id":"openstack%2Freleases~master~I520d75f39ef5f458e085df3f56573af68258811f","project":"openstack/releases","branch":"master","topic":"vitrage-approval","hashtags":[],"change_id":"I520d75f39ef5f458e085df3f56573af68258811f","subject":"Add Eyal\u0027s review address to Vitrage liaisons","status":"MERGED","created":"2020-02-13 10:24:09.000000000","updated":"2020-02-13 17:20:48.000000000","submitted":"2020-02-13 17:20:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707605-1581614448595-b9790512","meta_rev_id":"42465689ccc44714c1374db9a97fe751afacf0ed","_number":707605,"virtual_id_number":707605,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 17:20:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-13 17:01:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":2,"date":"2020-02-13 16:37:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-13 17:01:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 15:44:07.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-13 16:37:24.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-02-13 17:01:30.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-13 17:20:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"635b37d5bdc40be92e223c14ecd9688c68aa29ec","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-13 10:24:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1dec4c7e977be14d336e7bc77095970a46fb205f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 10:43:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62b33b4fcb5e42028798586597f3b80d : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3889b8ec538453fa0c385967f9f5e05 : SUCCESS in 14m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/70b69dee17b044d38aaf8aafc581667b : SUCCESS in 9m 26s","accounts_in_message":[],"_revision_number":1},{"id":"ccd9c0c5c370e498728ab2a0284eee3f8e4e45ec","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-13 15:44:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b80ae32a19f7e1c0cfabcfeab8cdd72e65513842","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-13 16:11:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nGood address","accounts_in_message":[],"_revision_number":1},{"id":"5d9f0ee01e4daae033f855560d973ee3b8c1f6e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:12:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1572d7c809bf29f90dbd6f6cfb7e2612de8fe45","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-13 16:28:00.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c18a602dbd77848416c334eebfaf5e3eb3f236a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:28:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a160eb498e74e55a2ee92efc696e9a5 : SUCCESS in 8m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae7b861bff5b4a159761f9021110700c : SUCCESS in 14m 35s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/0221b4ec8ca14837827927d0b0135183 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":1},{"id":"842a715746d35b7078d0d2a70e3be78a846ca298","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-13 16:30:21.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c96330cd48fb2bdfbd06dcaf61f4d70c3a7ae72e","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-13 16:37:24.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"042506685a362bbd7af4d62ef26eefc586a5298e","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-13 16:41:49.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nFor my blindness, I explained things with more words.","accounts_in_message":[],"_revision_number":2},{"id":"3a4b8901ff7d106881ef13c2324eea5b61e675aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:45:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b4d379309024b55ad450a525a6bd369 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88977105a81247b5a89b17dc02f434ec : SUCCESS in 13m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/58436b464a7a41b9aafa1b0f003a798b : SUCCESS in 8m 39s","accounts_in_message":[],"_revision_number":2},{"id":"57ac2300373cb6d4cacb6b34e07bead01f3752fe","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-13 17:01:30.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"abf65da712013d04a06fc24a1b1adfbc0b8d5188","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:01:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ab239cceebdd6dbdf5cffe3f797aec855c5f56e6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:20:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"68bf6ab7c2ac88d495e8f201ba1cb5e0bf02807d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:20:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb20689a52194ccfb550c2fcef559737 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dd7f7bd03e9468699a9a978518b261a : SUCCESS in 13m 07s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/8b538d6dd2854ca391656d87b3042325 : SUCCESS in 6m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6b4a656d8e935f60a57d24a82f51ec360fc22d84","revisions":{"0732477fc1342e47516e7e1d2dfe96abda4b5c5d":{"kind":"REWORK","_number":1,"created":"2020-02-13 10:24:09.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/05/707605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/707605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/707605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/707605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/707605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/707605/1"}}},"commit":{"parents":[{"commit":"bdc1fa7c58d52b01eda6af781ff40d45aeea6187","subject":"Merge \"Release sushy 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bdc1fa7c58d52b01eda6af781ff40d45aeea6187"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-13 10:22:33.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-13 10:22:33.000000000","tz":60},"subject":"Add Eyal\u0027s review address to Vitrage liaisons","message":"Add Eyal\u0027s review address to Vitrage liaisons\n\nEyal Bar-Ilan is using a different address for reviews than\nthe one defined in governance. Add the review address to\nthe Vitrage release liaisons so that check-ptl-approval\nworks as intended.\n\nChange-Id: I520d75f39ef5f458e085df3f56573af68258811f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0732477fc1342e47516e7e1d2dfe96abda4b5c5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0732477fc1342e47516e7e1d2dfe96abda4b5c5d"}]},"branch":"refs/heads/master"},"6b4a656d8e935f60a57d24a82f51ec360fc22d84":{"kind":"REWORK","_number":2,"created":"2020-02-13 16:30:21.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/05/707605/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/05/707605/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/707605/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/707605/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/05/707605/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/05/707605/2"}}},"commit":{"parents":[{"commit":"bdc1fa7c58d52b01eda6af781ff40d45aeea6187","subject":"Merge \"Release sushy 3.1.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bdc1fa7c58d52b01eda6af781ff40d45aeea6187"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-13 10:22:33.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-13 16:30:00.000000000","tz":60},"subject":"Add Eyal\u0027s review address to Vitrage liaisons","message":"Add Eyal\u0027s review address to Vitrage liaisons\n\nEyal Bar-Ilan is using a different address for reviews than\nthe one defined in governance. Add the review address to\nthe Vitrage release liaisons so that check-ptl-approval\nworks as intended.\n\nChange-Id: I520d75f39ef5f458e085df3f56573af68258811f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6b4a656d8e935f60a57d24a82f51ec360fc22d84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6b4a656d8e935f60a57d24a82f51ec360fc22d84"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
