)]}'
{"id":"openstack%2Freleases~707642","triplet_id":"openstack%2Freleases~master~I01a4c1121192e3dd022c0235a3429e08e6437149","project":"openstack/releases","branch":"master","hashtags":[],"change_id":"I01a4c1121192e3dd022c0235a3429e08e6437149","subject":"Release vitrage-dashboard 3.0.0 for Ussuri","status":"MERGED","created":"2020-02-13 14:11:47.000000000","updated":"2020-02-13 17:20:49.000000000","submitted":"2020-02-13 17:20:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707642-1581614449656-798fc2db","meta_rev_id":"09bf95dbe45ac93fe5951e70be7c18476239a024","_number":707642,"virtual_id_number":707642,"owner":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 17:20:49.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-13 17:01:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-02-13 16:30:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":2,"date":"2020-02-13 16:30:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-13 17:01:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-02-13 16:30:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 15:11:13.000000000","updated_by":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"reviewer":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"state":"REVIEWER"},{"updated":"2020-02-13 15:48:20.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-13 17:01:50.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-13 17:20:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d2dd572db37373ea641d8a1c16cf04759c7570ad","author":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"date":"2020-02-13 14:11:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b94750164886e93e6ca2bb0db0d55562b52225ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:24:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac56a58cc19c4f3394ec9d6cace06a28 : SUCCESS in 12m 21s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ce21031911d54a1893922ccd65bd3b63 : SUCCESS in 6m 18s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ab938343927e4a5094003ffc1b32f59a : SUCCESS in 5m 35s","accounts_in_message":[],"_revision_number":1},{"id":"43e5e76e1b3b063b29a8b27c7922d05c6b227119","author":{"_account_id":1736,"name":"Ivan Kolodyazhny","email":"e0ne@e0ne.info","username":"e0ne"},"date":"2020-02-13 15:11:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"511e1429271882c08aa1518cce76ae595f6b7baa","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-13 15:48:20.000000000","message":"Patch Set 1: Code-Review+2\n\nPy2 support dropped [1], release a major version, LGTM\n\n[1] https://github.com/openstack/vitrage-dashboard/commit/c9fb559b572334a2514a791aa931f5ea97b3be85","accounts_in_message":[],"_revision_number":1},{"id":"13d79982947c4e140e065068ae4d79f92f60cbd5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-13 16:01:16.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a4db15e2672b3d5276a948aa03229d1754b18caa","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:03:08.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"c84fc80bff5e7c5a82369afcf31595bac27e788d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:18:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1678df3bd99447dd9e02273257c1ca09 : SUCCESS in 12m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/23a1d77b7a8d40348813c266994aab12 : SUCCESS in 7m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/2b6578ca099343418ac93b13ef309e26 : FAILURE in 6m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5181988ca589441fd33af5abd3de308c71edf0d8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-13 16:30:55.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ac5edf4e1074fa3e4bf1cde8a4b56ff60d254937","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 16:46:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12883a40544d4faaa0e54b9ee8e0a61d : SUCCESS in 11m 19s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3e7f868f71234256a12ca23ea62ec9cc : SUCCESS in 8m 22s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/46e1f45303ec49aab66d5ecf0eb34774 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":3},{"id":"2aad30c9119819f2ff6f0131b59e5b5af6879624","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-13 17:01:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a35fe80d1f67d6e6ce6dc6bcab756ea4ef0a7973","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:02:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"dbde8924f742f1edce238cd9c89e4c4c73bb4c64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:20:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"09bf95dbe45ac93fe5951e70be7c18476239a024","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 17:20:49.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0a0d837ac254972855a92ecfb6d4538 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94c6a153cb1744e1a124cdefe22d72c6 : SUCCESS in 12m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/61c9bcdf83a64135a6ebd91042a8c2fa : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c0a7172778601d0f8f4e40dc50c1a31b31fc7584","revisions":{"2e3666e592d0941b5b53414a287c4842bcb7407b":{"kind":"REWORK","_number":1,"created":"2020-02-13 14:11:47.000000000","uploader":{"_account_id":19134,"name":"Eyal","email":"eyalb1@gmail.com","username":"eyalb"},"ref":"refs/changes/42/707642/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/707642/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/707642/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/707642/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/707642/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/707642/1"}}},"commit":{"parents":[{"commit":"4593872028ead152c4c549b3e9b26d4efed3a5db","subject":"Merge \"Release vitrage 6.0.0 for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4593872028ead152c4c549b3e9b26d4efed3a5db"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-13 14:09:16.000000000","tz":120},"committer":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-13 14:11:20.000000000","tz":120},"subject":"Release vitrage-dashboard 3.0.0 for Ussuri","message":"Release vitrage-dashboard 3.0.0 for Ussuri\n\nChange-Id: I01a4c1121192e3dd022c0235a3429e08e6437149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e3666e592d0941b5b53414a287c4842bcb7407b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2e3666e592d0941b5b53414a287c4842bcb7407b"}]},"branch":"refs/heads/master"},"ba3a0f01b2682c750a18c8d5748c9b15b09911ee":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-02-13 16:01:16.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/42/707642/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/707642/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/707642/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/707642/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/707642/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/707642/2"}}},"commit":{"parents":[{"commit":"0732477fc1342e47516e7e1d2dfe96abda4b5c5d","subject":"Add Eyal\u0027s review address to Vitrage liaisons","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0732477fc1342e47516e7e1d2dfe96abda4b5c5d"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-13 14:09:16.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-13 16:01:16.000000000","tz":0},"subject":"Release vitrage-dashboard 3.0.0 for Ussuri","message":"Release vitrage-dashboard 3.0.0 for Ussuri\n\nChange-Id: I01a4c1121192e3dd022c0235a3429e08e6437149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba3a0f01b2682c750a18c8d5748c9b15b09911ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ba3a0f01b2682c750a18c8d5748c9b15b09911ee"}]},"branch":"refs/heads/master"},"c0a7172778601d0f8f4e40dc50c1a31b31fc7584":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-02-13 16:30:55.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/42/707642/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/707642/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/707642/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/707642/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/707642/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/707642/3"}}},"commit":{"parents":[{"commit":"6b4a656d8e935f60a57d24a82f51ec360fc22d84","subject":"Add Eyal\u0027s review address to Vitrage liaisons","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6b4a656d8e935f60a57d24a82f51ec360fc22d84"}]}],"author":{"name":"Eyal","email":"eyal.bar-ilan@nokia.com","date":"2020-02-13 14:09:16.000000000","tz":120},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-13 16:30:55.000000000","tz":0},"subject":"Release vitrage-dashboard 3.0.0 for Ussuri","message":"Release vitrage-dashboard 3.0.0 for Ussuri\n\nChange-Id: I01a4c1121192e3dd022c0235a3429e08e6437149\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c0a7172778601d0f8f4e40dc50c1a31b31fc7584"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c0a7172778601d0f8f4e40dc50c1a31b31fc7584"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
