)]}'
{"id":"openstack%2Freleases~707651","triplet_id":"openstack%2Freleases~master~I5524048c0b318d03038431cacde558b0df0cb13d","project":"openstack/releases","branch":"master","topic":"oslo_drop_py2_support","hashtags":[],"change_id":"I5524048c0b318d03038431cacde558b0df0cb13d","subject":"Release oslo.i18n 4.0.0","status":"MERGED","created":"2020-02-13 14:30:27.000000000","updated":"2020-02-13 19:32:59.000000000","submitted":"2020-02-13 19:32:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"707651-1581622379011-a5ae4dff","meta_rev_id":"7e4fdfc83eb73534357ffee66ea2b4f63eefa1ed","_number":707651,"virtual_id_number":707651,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-13 19:32:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-13 19:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-13 19:13:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-02-13 19:15:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-13 19:13:46.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-13 19:32:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ddcce621b272f5d4bdc4bfb58d71f04854943d3e","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-13 14:30:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6580e25cfd07bd819a9699be4329c4a7eb211919","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-13 14:30:49.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"c39e5303cc8852f067c187775f200b24be0f2503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 14:45:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b7ff1e780ac4b1898430df84c595860 : SUCCESS in 12m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/aa3d6f83559743e2bbffb47590753485 : SUCCESS in 9m 02s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9327573c5d964d5c8edc58ea91ea1281 : SUCCESS in 6m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ddd2deaee84a05fd1e09208456f7ef5fc8fa863a","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-13 19:13:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e58d27751c5fefd3fb237febec08bd682530bc23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 19:14:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b400237a4b9ccc57a4446a6c9a942c525d7f2172","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 19:15:57.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"61ea78406c25bda56708e335f77aaf5a3e846a44","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 19:32:58.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bb93e102f624997a24063e7ef4a2aad : SUCCESS in 7m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f5b87770ab45cc86b03548ebb756ba : SUCCESS in 13m 30s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb74518bcb3341c7b0e0b903ccc1b8f4 : SUCCESS in 8m 38s","accounts_in_message":[],"_revision_number":2},{"id":"2700906452e7678c26758c0b3989794d3f78c27a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 19:32:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"93d32d86a3db12d5760cf3267ce9f6fdc1632bac","revisions":{"9c8f8923a3a66ce1ba74eddd152e1a4b726d1f83":{"kind":"REWORK","_number":1,"created":"2020-02-13 14:30:27.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/707651/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/707651/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/707651/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/707651/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/707651/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/707651/1"}}},"commit":{"parents":[{"commit":"4593872028ead152c4c549b3e9b26d4efed3a5db","subject":"Merge \"Release vitrage 6.0.0 for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4593872028ead152c4c549b3e9b26d4efed3a5db"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-13 14:29:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-13 14:29:54.000000000","tz":60},"subject":"Release oslo.i18n 2.0.0","message":"Release oslo.i18n 2.0.0\n\nThis version officially drop the python 2 support so we release a major\nversion of oslo.i18n to inform user that we break\nthe backward compatibility.\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n-\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: I5524048c0b318d03038431cacde558b0df0cb13d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9c8f8923a3a66ce1ba74eddd152e1a4b726d1f83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/9c8f8923a3a66ce1ba74eddd152e1a4b726d1f83"}]},"branch":"refs/heads/master"},"93d32d86a3db12d5760cf3267ce9f6fdc1632bac":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-13 14:30:49.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/51/707651/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/51/707651/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/707651/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/707651/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/51/707651/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/51/707651/2"}}},"commit":{"parents":[{"commit":"4593872028ead152c4c549b3e9b26d4efed3a5db","subject":"Merge \"Release vitrage 6.0.0 for Ussuri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4593872028ead152c4c549b3e9b26d4efed3a5db"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-13 14:29:54.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-13 14:30:33.000000000","tz":60},"subject":"Release oslo.i18n 4.0.0","message":"Release oslo.i18n 4.0.0\n\nThis version officially drop the python 2 support so we release a major\nversion of oslo.i18n to inform user that we break\nthe backward compatibility.\n\nOpenStack is dropping the py2.7 support in ussuri cycle.\n\nComplete discussion \u0026 schedule can be found in\n-\nhttp://lists.openstack.org/pipermail/openstack-discuss/2019-October/010142.html\n- https://etherpad.openstack.org/p/drop-python2-support\n\nUssuri Communtiy-wide goal:\nhttps://governance.openstack.org/tc/goals/selected/ussuri/drop-py27.html\n\nChange-Id: I5524048c0b318d03038431cacde558b0df0cb13d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/93d32d86a3db12d5760cf3267ce9f6fdc1632bac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/93d32d86a3db12d5760cf3267ce9f6fdc1632bac"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
