)]}'
{"id":"openstack%2Freleases~707875","triplet_id":"openstack%2Freleases~master~I66da5156816c9a8113cb0faefe6756e3d41017a2","project":"openstack/releases","branch":"master","topic":"victoria","hashtags":[],"change_id":"I66da5156816c9a8113cb0faefe6756e3d41017a2","subject":"Proposed release schedule for Victoria (21w)","status":"ABANDONED","created":"2020-02-14 17:11:16.000000000","updated":"2020-02-17 13:50:17.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ec30a79ceb650568fe9329aa9dd32ab15355b500","_number":707875,"virtual_id_number":707875,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-02-17 10:22:18.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-14 17:29:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-14 17:29:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-17 10:22:18.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"5378220ff84f191bc37ab84afab1f75434b4ceae","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-14 17:11:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b4a717b30a035cc9e7841294fe4d07816abd9def","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-14 17:15:55.000000000","message":"Topic changed from victoria-20 to victoria","accounts_in_message":[],"_revision_number":1},{"id":"021c3630379131099134c84c26810941f04e21e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 17:29:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f347baf2c16f4c589013335ff354c896 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30cf42df383f4aabbdf0307f5422d9a5 : SUCCESS in 13m 04s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/76718e0a876b46ddb799a22283e467bc : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":1},{"id":"97877e1de9cd28e9f76fd413020cc0ceb82b68a8","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-17 10:22:18.000000000","message":"Patch Set 1:\n\n21 weeks is short, but not the shortest we\u0027ve ever made (was 20 weeks for Ocata). If we pick this one, I would balance it a bit better by placing milestones on Jun 18 and July 23 (5+5+6 instead of 4+5+7).","accounts_in_message":[],"_revision_number":1},{"id":"ec30a79ceb650568fe9329aa9dd32ab15355b500","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-17 13:50:17.000000000","message":"Abandoned\n\nThrew noodle on wall, did not stick.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37d9ac62e12b414a768227244677353f037cb9c9","revisions":{"37d9ac62e12b414a768227244677353f037cb9c9":{"kind":"REWORK","_number":1,"created":"2020-02-14 17:11:16.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/75/707875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/707875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/707875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/707875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/707875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/707875/1"}}},"commit":{"parents":[{"commit":"4f589277950c6896a41414f152c09f17a99751c0","subject":"Merge \"[cinder] Ussuri milestone-2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f589277950c6896a41414f152c09f17a99751c0"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-14 16:08:29.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-14 16:46:54.000000000","tz":-360},"subject":"Proposed release schedule for Victoria (21w)","message":"Proposed release schedule for Victoria (21w)\n\nThis proposes a release schedule for Victoria that lasts 21 weeks to end\ntwo weeks before the fall event.\n\nChange-Id: I66da5156816c9a8113cb0faefe6756e3d41017a2\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/37d9ac62e12b414a768227244677353f037cb9c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/37d9ac62e12b414a768227244677353f037cb9c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
