)]}'
{"id":"openstack%2Freleases~707876","triplet_id":"openstack%2Freleases~master~I9b5b8ffef2b65a4d9f1a587ff76b13dcdd3502ee","project":"openstack/releases","branch":"master","topic":"victoria","hashtags":[],"change_id":"I9b5b8ffef2b65a4d9f1a587ff76b13dcdd3502ee","subject":"Proposed release schedule for Victoria (23w)","status":"ABANDONED","created":"2020-02-14 17:11:32.000000000","updated":"2020-02-21 14:34:05.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6a59185415e94f14453ec371668c46d401390554","_number":707876,"virtual_id_number":707876,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-02-18 16:27:18.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-17 14:24:12.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-02-18 17:44:17.000000000","_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-02-20 13:08:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-02-20 16:05:56.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"date":"2020-02-17 20:15:00.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 14:24:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-17 20:15:00.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-02-18 17:44:17.000000000","updated_by":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"reviewer":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"state":"REVIEWER"},{"updated":"2020-02-20 13:08:50.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-02-20 16:05:56.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"}],"messages":[{"id":"33d6decdd72c4b68e52bfe708b7ec1002316292c","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-14 17:11:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f2f10217d1c2341fa580bf488c7da140932f379d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 17:24:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c90e57c0ce164ff8a6a053db9da7b93d : SUCCESS in 8m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ded7a0ff8dfc4fe99630608782177909 : SUCCESS in 12m 15s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/40a9e85d63da4b04981f6c347c293ded : SUCCESS in 6m 12s","accounts_in_message":[],"_revision_number":1},{"id":"e0a0b3c30d78d9d0acd651b8436a93971bcbf844","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-17 10:24:53.000000000","message":"Patch Set 1:\n\nI like this one... but I wonder if it would not make better sense to do the release the week of the event (rather than the week after, which would be a bit awkward in terms of announcements).\n\nSo basically do 23w with milestone on Sep 17. WDYT?","accounts_in_message":[],"_revision_number":1},{"id":"cb81ee12893e4307b478affac18bc714dc4c559e","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-17 14:04:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b1427eb7fbe48a9c74563df124bc34a01bfbfb43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 14:24:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6eb69e1cc7c04c70b0d57638efd822ec : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af9ecde1f99c4e49a49febe586016330 : SUCCESS in 13m 08s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d91ea625d79a47c6830f046d248115be : SUCCESS in 6m 58s","accounts_in_message":[],"_revision_number":2},{"id":"5fb1a064de0a01ed47bfc6d911994307f80dbdc5","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-17 16:24:11.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM\n\nI agree with Thierry it make more sense to base release date on event.\n\nAlso I think 23w is a good compromize between other proposed release schedules (21w, 27w).\n\nI suppose we must +2 if we agree with this model, so I +2.","accounts_in_message":[],"_revision_number":2},{"id":"90fb261ddb0a5f9c80e9b2063de78e634e117b20","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-18 10:30:17.000000000","message":"Patch Set 2: Code-Review+2\n\nOK to propose for wider feedback on the ML","accounts_in_message":[],"_revision_number":2},{"id":"b6bd9008288ea6265f13b6113a26cf2462a5bb51","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-02-18 16:24:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c3c25f9eceb827754d244ec90810789ac27ad9a2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-18 16:27:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"76747c9da911c76c868fb2dc903c2b101338371d","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-02-18 17:01:08.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3d842b9b83e02467ade60befedb83cec803c2864","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-18 17:27:35.000000000","message":"Patch Set 2:\n\n@Corey: It makes a short cycle slightly shorter, but would also make the busy summit week a bit lighter, so I think that would work equally well for me.","accounts_in_message":[],"_revision_number":2},{"id":"3b13058a7f126aa6edacb1f8d57aff4869d8d418","author":{"_account_id":11805,"name":"Corey Bryant","email":"corey.bryant@canonical.com","username":"coreycb"},"date":"2020-02-18 17:44:17.000000000","message":"Patch Set 2:\n\n@thierry, Great, yes releasing on summit week sounds exciting but not ideal for those involved anyway. :)","accounts_in_message":[],"_revision_number":2},{"id":"094259777bb21d1bdc54a8f4d7c2f08a0c6a8e37","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-20 13:08:50.000000000","message":"Patch Set 2: -Code-Review\n\nI would be fine with this one but prefer https://review.opendev.org/#/c/708471/1 which should make it easier on release team and Canonical.","accounts_in_message":[],"_revision_number":2},{"id":"89e75de4b42cdd3f8f72a1e8e4c7f6c35b80623f","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-20 16:05:56.000000000","message":"Patch Set 2: -Code-Review\n\n\u003e I would be fine with this one but prefer https://review.opendev.org/#/c/708471/1\n \u003e which should make it easier on release team and Canonical.\n\nSame things for me.","accounts_in_message":[],"_revision_number":2},{"id":"8f00ca0bdda76dce9cc53d4625a431c5daa2a59b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-21 14:34:05.000000000","message":"Abandoned\n\nGoing with the 22 week proposal since it aligns better.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"81b791c62ff6bac3d6cb61ccffa4c564d42ad598","revisions":{"222481bf014cf113565988c2662e9c80d2403ab8":{"kind":"REWORK","_number":1,"created":"2020-02-14 17:11:32.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/76/707876/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/707876/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/707876/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/707876/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/707876/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/707876/1"}}},"commit":{"parents":[{"commit":"4f589277950c6896a41414f152c09f17a99751c0","subject":"Merge \"[cinder] Ussuri milestone-2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f589277950c6896a41414f152c09f17a99751c0"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-14 16:44:39.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-14 17:10:44.000000000","tz":-360},"subject":"Proposed release schedule for Victoria (24w)","message":"Proposed release schedule for Victoria (24w)\n\nThis proposes a 24 week release schedule to align the release the week\nafter the fall Summit event.\n\nChange-Id: I9b5b8ffef2b65a4d9f1a587ff76b13dcdd3502ee\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/222481bf014cf113565988c2662e9c80d2403ab8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/222481bf014cf113565988c2662e9c80d2403ab8"}]},"branch":"refs/heads/master"},"81b791c62ff6bac3d6cb61ccffa4c564d42ad598":{"kind":"REWORK","_number":2,"created":"2020-02-17 14:04:18.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/76/707876/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/76/707876/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/707876/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/707876/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/76/707876/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/76/707876/2"}}},"commit":{"parents":[{"commit":"4f589277950c6896a41414f152c09f17a99751c0","subject":"Merge \"[cinder] Ussuri milestone-2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f589277950c6896a41414f152c09f17a99751c0"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-14 16:44:39.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-17 14:03:54.000000000","tz":-360},"subject":"Proposed release schedule for Victoria (23w)","message":"Proposed release schedule for Victoria (23w)\n\nThis proposes a 23 week release schedule to align the release the week\nof the fall Summit event.\n\nChange-Id: I9b5b8ffef2b65a4d9f1a587ff76b13dcdd3502ee\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81b791c62ff6bac3d6cb61ccffa4c564d42ad598"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/81b791c62ff6bac3d6cb61ccffa4c564d42ad598"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
