)]}'
{"id":"openstack%2Freleases~707877","triplet_id":"openstack%2Freleases~master~Iaae7aeb40aa4ab86a9c88baa5e3c0763544521fe","project":"openstack/releases","branch":"master","topic":"victoria","hashtags":[],"change_id":"Iaae7aeb40aa4ab86a9c88baa5e3c0763544521fe","subject":"Proposed release schedule for Victoria (27w)","status":"ABANDONED","created":"2020-02-14 17:11:45.000000000","updated":"2020-02-17 13:50:38.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"943257a0c6bd4ef05e13cc78b52251ae5d82dac6","_number":707877,"virtual_id_number":707877,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-02-17 10:26:18.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-14 17:28:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"permitted_voting_range":{"min":0,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-14 17:28:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-17 10:26:18.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"}],"messages":[{"id":"52e0130dc80915c4ae94b8f07de6fcd41bb85c40","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-14 17:11:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d8120ef21555c28932c938197c7bec7845216e2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-14 17:16:03.000000000","message":"Topic changed from victoria2 to victoria","accounts_in_message":[],"_revision_number":1},{"id":"16886c4175bf10dff840ffdf7dd353f5e9cd2d77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-14 17:28:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcbecc3202d14e03b2b7d6308bd78b25 : SUCCESS in 7m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c816d90c5bf84f28ab941c629770bf96 : SUCCESS in 12m 16s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ca8c72031d3a422a88614547f51854c5 : SUCCESS in 6m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b578b61170feb330b8577e67691b7a691ef8c273","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-17 10:26:18.000000000","message":"Patch Set 1:\n\nNot a big fan of this one... I fear the disruption of having summit travel the week of milestone-3 and then disrupt RC1 generation by having summit in the middle.","accounts_in_message":[],"_revision_number":1},{"id":"943257a0c6bd4ef05e13cc78b52251ae5d82dac6","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-17 13:50:38.000000000","message":"Abandoned\n\nNot ideal.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eb9d45db1ebdaab5d42e20d73182cf1a53daf143","revisions":{"eb9d45db1ebdaab5d42e20d73182cf1a53daf143":{"kind":"REWORK","_number":1,"created":"2020-02-14 17:11:45.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/77/707877/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/77/707877/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/707877/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/707877/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/77/707877/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/77/707877/1"}}},"commit":{"parents":[{"commit":"4f589277950c6896a41414f152c09f17a99751c0","subject":"Merge \"[cinder] Ussuri milestone-2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4f589277950c6896a41414f152c09f17a99751c0"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-14 17:09:09.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-14 17:09:09.000000000","tz":-360},"subject":"Proposed release schedule for Victoria (27w)","message":"Proposed release schedule for Victoria (27w)\n\nThis is a proposed 27 week schedule for Victoria so the Summit falls\nnear the time when development activity is winding down.\n\nChange-Id: Iaae7aeb40aa4ab86a9c88baa5e3c0763544521fe\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb9d45db1ebdaab5d42e20d73182cf1a53daf143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/eb9d45db1ebdaab5d42e20d73182cf1a53daf143"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
