)]}'
{"id":"openstack%2Freleases~708065","triplet_id":"openstack%2Freleases~master~I8cbe0e90dc6c53d4513a03f5e03c8e5926c73d70","project":"openstack/releases","branch":"master","topic":"puppet-openstack-ussuri-m2","hashtags":[],"change_id":"I8cbe0e90dc6c53d4513a03f5e03c8e5926c73d70","subject":"Puppet OpenStack Ussuri Milestone 2","status":"MERGED","created":"2020-02-17 07:57:25.000000000","updated":"2020-02-18 10:56:29.000000000","submitted":"2020-02-18 10:56:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":43,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708065-1582023389152-86a07e3f","meta_rev_id":"0fe32c21bea8942b8980a87655636fb5775fea92","_number":708065,"virtual_id_number":708065,"owner":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-02-18 02:37:56.000000000","_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-18 10:56:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":2,"date":"2020-02-18 10:33:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":1,"date":"2020-02-18 07:26:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"all":[{"value":1,"date":"2020-02-18 10:33:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-02-17 15:33:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 07:26:36.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-18 10:33:20.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-02-18 10:56:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f7e659d3d88062993de86cbecf5db04170253956","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-02-17 07:57:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"10753cd274b8d65e247ff4c98409549fbde5435f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 08:11:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d7bf651966409fb8ede3a9370de19b : SUCCESS in 12m 45s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d38292d5bc564629a401840acbed370a : SUCCESS in 12m 42s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3ab5f8e1022d491f86b71954996442a9 : SUCCESS in 7m 53s","accounts_in_message":[],"_revision_number":1},{"id":"07ba72e36f2b0fc474fbec3bdcfb6447ac094516","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-17 15:32:02.000000000","message":"Patch Set 1: Code-Review-1\n\n(43 comments)\n\nHello,\n\nMany projects shouldn\u0027t release a minor version, see my inline comments.","accounts_in_message":[],"_revision_number":1},{"id":"7e9ad154e3d84fe0470f3ddce30444704d734d70","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-17 15:33:24.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"5b2450d7414d5d1e38e261dd98205eafcb0e96ff","author":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"date":"2020-02-18 02:37:56.000000000","message":"Patch Set 1:\n\nThe version of the puppet openstack project is different from that of other projects. The version must be the same as that in the metadata.json file.","accounts_in_message":[],"_revision_number":1},{"id":"4d7871119f8b8317e1e77fe1aca1949909938d71","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-18 07:26:36.000000000","message":"Patch Set 1: Code-Review+1\n\n\u003e The version of the puppet openstack project is different from that\n \u003e of other projects. The version must be the same as that in the\n \u003e metadata.json file.\n\nAh ok I was not aware of this rule, it\u0027s a little bit annoying because many releases does not really reflect associated changes\n\nConsidering the huge amount of changes needed to address my comments I don\u0027t think that this is something that you want to consider, I\u0027m right?\n\nSince you are the PTL/liaison of this team, I think we are \"safe\" here, so I remove my -1 and move to +1 only for now by waiting other team release member feedbacks and if they will be okay with these changes too then I\u0027ll +2 too. In other words looks good to me, but someone else must approve.","accounts_in_message":[],"_revision_number":1},{"id":"e33322aa8c867db264303f8c8551a647d30a37b5","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-18 10:33:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"883a5ac529963db6992b2db5b1aeb8a31e27adf3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 10:33:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5adc0f3d27dcabb63e64143181d050008403b36c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 10:56:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf63a80847cd4afc9f3c0e922996d1bb : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6721ae6982544c39bf19a34e60ff6d2b : SUCCESS in 14m 36s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/cb500f89ca5541d4bd4131da8b2918b1 : SUCCESS in 17m 10s","accounts_in_message":[],"_revision_number":1},{"id":"bf81cadc56b7283c63ef56f64eb066a20dbc9676","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 10:56:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fe40c00c3ef95f729806f98097cbb3ead363f2a2","revisions":{"fe40c00c3ef95f729806f98097cbb3ead363f2a2":{"kind":"REWORK","_number":1,"created":"2020-02-17 07:57:25.000000000","uploader":{"_account_id":9414,"name":"zhongshengping","email":"chdzsp@163.com","username":"ZhongShengping"},"ref":"refs/changes/65/708065/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/708065/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/708065/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/708065/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/708065/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/708065/1"}}},"commit":{"parents":[{"commit":"6d5cfbf1ab5e1cee2189c13be57f704331ee889a","subject":"Merge \"[zaqar] Ussuri milestone-2 release\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/6d5cfbf1ab5e1cee2189c13be57f704331ee889a"}]}],"author":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-02-17 07:52:25.000000000","tz":480},"committer":{"name":"ZhongShengping","email":"chdzsp@163.com","date":"2020-02-17 07:56:31.000000000","tz":480},"subject":"Puppet OpenStack Ussuri Milestone 2","message":"Puppet OpenStack Ussuri Milestone 2\n\nChange-Id: I8cbe0e90dc6c53d4513a03f5e03c8e5926c73d70\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fe40c00c3ef95f729806f98097cbb3ead363f2a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/fe40c00c3ef95f729806f98097cbb3ead363f2a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
