)]}'
{"id":"openstack%2Freleases~708623","triplet_id":"openstack%2Freleases~master~I26ac918f66146c7b818e646ce54b2aea84a302ac","project":"openstack/releases","branch":"master","topic":"adrian-email","hashtags":[],"change_id":"I26ac918f66146c7b818e646ce54b2aea84a302ac","subject":"Add Adrian\u0027s review address to adjutant liaisons","status":"MERGED","created":"2020-02-19 11:42:55.000000000","updated":"2020-02-19 17:21:33.000000000","submitted":"2020-02-19 17:21:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708623-1582132893145-0cc33de8","meta_rev_id":"159b2abc47f2598b2469cad7f942f692b803ee60","_number":708623,"virtual_id_number":708623,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 17:21:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-19 17:03:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-19 17:03:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-19 17:03:20.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-19 17:21:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"914a2b6d14f106e3d7b1bcf9699fb25225a01b65","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-19 11:42:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b33c4e67866a43b94ef91b5737fac4d38661adf8","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-19 11:43:26.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3daef8c58998c40a6a9d047351a0b009b5b1fe41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 12:02:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0ff10f0653540c683226ed8fe5bb263 : SUCCESS in 5m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7294ded05921472bbeffebcfec8f987b : SUCCESS in 13m 20s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6a18dc3df784e4f8e38f69ad7301fe5 : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":2},{"id":"de5445eb0695b9301ccf1dfa4c166127914f6f94","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-19 17:03:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"8ce15f1d940103739de84bf117b9574e55efebaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 17:04:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0371ca1993b3bcdf1e0e6fe1f85f91aa0fc904a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 17:21:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10d80cb880944526b3b7f01420e911e2 : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2348d69b02774edb96fa935c8e77c74d : SUCCESS in 13m 25s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3ea3db664c66468ba642254d6a8b7839 : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":2},{"id":"60165889a95105af08b5ea2d3815a6244a9e892c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 17:21:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"b4106ae7f2da234d1102abf5cde24841dcf2c2e9","revisions":{"d9c1fb0e96a0f8375679447460f88e74524609b4":{"kind":"REWORK","_number":1,"created":"2020-02-19 11:42:55.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/23/708623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/708623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/708623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/708623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/708623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/708623/1"}}},"commit":{"parents":[{"commit":"13a87431215f15d9be0a68e5b9346b9da8ae8661","subject":"Merge \"Revert \"First release of sushy-cli: 0.1.0 for Ussuri\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/13a87431215f15d9be0a68e5b9346b9da8ae8661"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-19 11:41:45.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-19 11:41:45.000000000","tz":60},"subject":" Add Adrian\u0027s review address to adjutant liaisons","message":" Add Adrian\u0027s review address to adjutant liaisons\n\nAdrian Turjak is using a different address for reviews than\nthe one defined in governance. Add the review address to\nthe adjutant release liaisons so that check-ptl-approval\nworks as intended.\n\nChange-Id: I26ac918f66146c7b818e646ce54b2aea84a302ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9c1fb0e96a0f8375679447460f88e74524609b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d9c1fb0e96a0f8375679447460f88e74524609b4"}]},"branch":"refs/heads/master"},"b4106ae7f2da234d1102abf5cde24841dcf2c2e9":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-02-19 11:43:26.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/23/708623/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/23/708623/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/708623/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/708623/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/23/708623/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/23/708623/2"}}},"commit":{"parents":[{"commit":"13a87431215f15d9be0a68e5b9346b9da8ae8661","subject":"Merge \"Revert \"First release of sushy-cli: 0.1.0 for Ussuri\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/13a87431215f15d9be0a68e5b9346b9da8ae8661"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-19 11:41:45.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2020-02-19 11:43:14.000000000","tz":60},"subject":"Add Adrian\u0027s review address to adjutant liaisons","message":"Add Adrian\u0027s review address to adjutant liaisons\n\nAdrian Turjak is using a different address for reviews than\nthe one defined in governance. Add the review address to\nthe adjutant release liaisons so that check-ptl-approval\nworks as intended.\n\nChange-Id: I26ac918f66146c7b818e646ce54b2aea84a302ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4106ae7f2da234d1102abf5cde24841dcf2c2e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/b4106ae7f2da234d1102abf5cde24841dcf2c2e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
