)]}'
{"id":"openstack%2Freleases~709165","triplet_id":"openstack%2Freleases~master~I213a387ec5d262d85790425c1a1b0f5da4134d10","project":"openstack/releases","branch":"master","topic":"rocky-final","hashtags":[],"change_id":"I213a387ec5d262d85790425c1a1b0f5da4134d10","subject":"stable/rocky releases for keystone deliverables","status":"MERGED","created":"2020-02-21 15:59:46.000000000","updated":"2020-02-24 19:06:12.000000000","submitted":"2020-02-24 19:06:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709165-1582571172709-dfe5b5ea","meta_rev_id":"e684224ae305c3f0c8475d7e10674feb9ab28788","_number":709165,"virtual_id_number":709165,"owner":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-24 19:06:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-24 18:47:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":2,"date":"2020-02-24 14:28:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-24 18:47:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-24 14:28:14.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-24 18:47:35.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-24 19:06:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6cae7ee57aa05f47033f3e5c0384cba4e16741a8","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-02-21 15:59:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fb13e65c5e0fac6def22f9301ce71716de8ae72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 16:14:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9db22bad0d248adab8e492962c66bac : SUCCESS in 13m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/9a7f0370275449f889eae9b7d65accad : FAILURE in 11m 08s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d51b8639850048998482b4292aaabf90 : SUCCESS in 9m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b861b074dff2ff49d4ff718cb6f6c57efb5bb688","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-02-21 16:25:47.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bd525412b33b6c46402f51b2c50c973b3b018278","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-02-21 16:52:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7920f5fc8994bccc49877a791e1fa7afc271c9a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-21 17:13:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dfc5301ed624e689f660bf5e40be3ac : SUCCESS in 15m 05s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/4a7a2f24df24422a8421989561fdb319 : FAILURE in 11m 51s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9feb9321ff974b609c287be34d0aa77a : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":2},{"id":"08afc8310f20deb36abc5e0dacf44f30f4e03abe","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-24 11:08:25.000000000","message":"Patch Set 2: Code-Review-1\n\nkeystone should ignore kilo branches for release notes first (cf. similar issue on ceilometer \u003d\u003e https://review.opendev.org/#/c/707425/)","accounts_in_message":[],"_revision_number":2},{"id":"ee0ef8895c062e9c47ac948c45a893679ccd2706","author":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"date":"2020-02-24 14:18:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"73193f7df705fb17737b176305a44d34c7435263","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-24 14:28:14.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM waiting for CI.\n\nIgnoring kilo branch is now done by pointing on \u003d\u003e https://github.com/openstack/keystone/commit/dbc64b48b5ac376c93cb70022698c5c87da17288","accounts_in_message":[],"_revision_number":3},{"id":"10225686966fb58677aaca1c6f5eefc82bbed210","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 14:39:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53b3ca1511694c81ac8493df84c9041c : SUCCESS in 15m 17s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/b8b6fc2bc0b44b7f8125d9ea9a83879b : SUCCESS in 14m 48s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/fd85877f99a14dd9a94dbf327f36bf50 : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":3},{"id":"13b6af843aee475086875210cc74f5e373a6ec75","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-24 18:47:35.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nIncluded changes look appropriate, and version bumps match content. Looks good to me.","accounts_in_message":[],"_revision_number":3},{"id":"089c0f4388b3e116dabbb1c7b3ffba2813a51b4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 18:47:56.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"cb8d918309e27e5b1ca74a2ecd2c0640de1ecaa0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 19:06:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7649a29485644a649909619462d408aa : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2274fe9ba5794aa0989b25c9a465c256 : SUCCESS in 14m 12s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5cc23c494dab49cbba1c57f6e2def52e : SUCCESS in 13m 21s","accounts_in_message":[],"_revision_number":3},{"id":"62a71512a6209227ca0a29b956a637402a0fd6f4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 19:06:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"1c2e5222cce8e24faa00ea0bbc02a94f610074cb","revisions":{"bbaef50134ec51d7487f0c46f94014f1b955d900":{"kind":"REWORK","_number":1,"created":"2020-02-21 15:59:46.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/65/709165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/709165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/709165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/709165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/709165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/709165/1"}}},"commit":{"parents":[{"commit":"4ec30c4e0ed22cd565083077c3d9e5362dfc565d","subject":"Merge \"Release os-brick 3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ec30c4e0ed22cd565083077c3d9e5362dfc565d"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2020-02-21 15:57:01.000000000","tz":-300},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2020-02-21 15:57:01.000000000","tz":-300},"subject":"stable/rocky releases for keystone deliverables","message":"stable/rocky releases for keystone deliverables\n\n- keystone 14.2.0\n- keystoneauth 3.11.1\n- keystonemiddleware 5.2.2\n\nThese are likely the last releases in rocky.\n\nChange-Id: I213a387ec5d262d85790425c1a1b0f5da4134d10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bbaef50134ec51d7487f0c46f94014f1b955d900"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/bbaef50134ec51d7487f0c46f94014f1b955d900"}]},"branch":"refs/heads/master"},"4de32d182643a0eef7c0fe430196797743ced263":{"kind":"REWORK","_number":2,"created":"2020-02-21 16:52:51.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/65/709165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/709165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/709165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/709165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/709165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/709165/2"}}},"commit":{"parents":[{"commit":"4ec30c4e0ed22cd565083077c3d9e5362dfc565d","subject":"Merge \"Release os-brick 3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ec30c4e0ed22cd565083077c3d9e5362dfc565d"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2020-02-21 15:57:01.000000000","tz":-300},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2020-02-21 16:46:56.000000000","tz":-300},"subject":"stable/rocky releases for keystone deliverables","message":"stable/rocky releases for keystone deliverables\n\n- keystone 14.2.0\n- keystoneauth 3.10.1\n- keystonemiddleware 5.2.2\n\nThese are likely the last releases in rocky.\n\nChange-Id: I213a387ec5d262d85790425c1a1b0f5da4134d10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4de32d182643a0eef7c0fe430196797743ced263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4de32d182643a0eef7c0fe430196797743ced263"}]},"branch":"refs/heads/master"},"1c2e5222cce8e24faa00ea0bbc02a94f610074cb":{"kind":"REWORK","_number":3,"created":"2020-02-24 14:18:16.000000000","uploader":{"_account_id":16465,"name":"Kristi Nikolla","email":"knikolla@bu.edu","username":"knikolla"},"ref":"refs/changes/65/709165/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/65/709165/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/709165/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/709165/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/65/709165/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/65/709165/3"}}},"commit":{"parents":[{"commit":"4ec30c4e0ed22cd565083077c3d9e5362dfc565d","subject":"Merge \"Release os-brick 3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/4ec30c4e0ed22cd565083077c3d9e5362dfc565d"}]}],"author":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2020-02-21 15:57:01.000000000","tz":-300},"committer":{"name":"Kristi Nikolla","email":"kristi@nikolla.me","date":"2020-02-24 14:17:59.000000000","tz":-300},"subject":"stable/rocky releases for keystone deliverables","message":"stable/rocky releases for keystone deliverables\n\n- keystone 14.2.0\n- keystoneauth 3.10.1\n- keystonemiddleware 5.2.2\n\nThese are likely the last releases in rocky.\n\nChange-Id: I213a387ec5d262d85790425c1a1b0f5da4134d10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1c2e5222cce8e24faa00ea0bbc02a94f610074cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/1c2e5222cce8e24faa00ea0bbc02a94f610074cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
