)]}'
{"id":"openstack%2Freleases~709742","triplet_id":"openstack%2Freleases~master~I3232435d95d20e7406dd9b7e091e9fca82668bd3","project":"openstack/releases","branch":"master","topic":"fix-lbaas-dashboard","hashtags":[],"change_id":"I3232435d95d20e7406dd9b7e091e9fca82668bd3","subject":"neutron-lbaas-dashboard is a deprecated deliverable","status":"MERGED","created":"2020-02-25 14:39:03.000000000","updated":"2020-02-25 20:08:17.000000000","submitted":"2020-02-25 20:08:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709742-1582661297001-d353a4dd","meta_rev_id":"b57b035f2134a8a03928158c837a2383e75593c5","_number":709742,"virtual_id_number":709742,"owner":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-25 20:08:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-25 14:49:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-25 19:53:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 19:53:48.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-25 20:08:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"30d17a26acc61b1b94d3e64936adcf5930ed79c2","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-02-25 14:39:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cab8fb7b61d4b33b014377b24790bea99eaedc2","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-25 14:49:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7428f8bd69167257f3dd6ed9b41fd5e415304286","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 14:53:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d47903041fad4b6aa950f771cb5736f7 : SUCCESS in 13m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d7b21409f4514db889536a459cbe64a4 : SUCCESS in 10m 04s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/42ee7fabd08745bcbc57140cfb80bb3a : SUCCESS in 8m 28s","accounts_in_message":[],"_revision_number":1},{"id":"e853dd11adb50f893ad35d418b15d868449c9df1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-25 19:53:48.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b87b06b3392874655726f50b35aa5b4c384d5f11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 19:54:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eeb3d356c46a6c3ea639de68e376a7fd4d356199","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 20:08:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1e4364c37784fc2a4d20b6db798cab8 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b3b70f0f5454536be41b8e5bf9e5485 : SUCCESS in 12m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ed3de06b191c4c14b04e89c286d1a015 : SUCCESS in 7m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a7cae10054bc7c5073a8b26f5028da964ad47884","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 20:08:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c1dbbfb3ac2313bbd638f12a213911f49b948564","revisions":{"c1dbbfb3ac2313bbd638f12a213911f49b948564":{"kind":"REWORK","_number":1,"created":"2020-02-25 14:39:03.000000000","uploader":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"ref":"refs/changes/42/709742/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/42/709742/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/709742/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/709742/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/42/709742/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/42/709742/1"}}},"commit":{"parents":[{"commit":"2c5254e33fb679b94663759e3532cfe5339898f3","subject":"Merge \"Oslo releases for 2020-02-24\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c5254e33fb679b94663759e3532cfe5339898f3"}]}],"author":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-25 14:37:32.000000000","tz":60},"committer":{"name":"Thierry Carrez","email":"thierry@openstack.org","date":"2020-02-25 14:37:32.000000000","tz":60},"subject":"neutron-lbaas-dashboard is a deprecated deliverable","message":"neutron-lbaas-dashboard is a deprecated deliverable\n\nSince https://review.opendev.org/#/c/705780/, neutron-lbaas-dashboard\nis no longer an active OpenStack deliverable. Align the release model\non its (independent) deliverable file accordingly.\n\nChange-Id: I3232435d95d20e7406dd9b7e091e9fca82668bd3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c1dbbfb3ac2313bbd638f12a213911f49b948564"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/c1dbbfb3ac2313bbd638f12a213911f49b948564"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
