)]}'
{"id":"openstack%2Freleases~709753","triplet_id":"openstack%2Freleases~master~Ic0fbcb34a617f4dec30643b8bc5211f40bf9108f","project":"openstack/releases","branch":"master","topic":"oslo_drop_py2_support","hashtags":[],"change_id":"Ic0fbcb34a617f4dec30643b8bc5211f40bf9108f","subject":"oslotest 4.0.0","status":"MERGED","created":"2020-02-25 15:19:08.000000000","updated":"2020-02-28 21:05:45.000000000","submitted":"2020-02-28 21:05:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709753-1582923945630-b16e9fe7","meta_rev_id":"11e12e915dfd2a51f159299e23a3d9eebd3c5b4a","_number":709753,"virtual_id_number":709753,"owner":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-02-27 22:02:55.000000000","_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-28 21:05:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2020-02-28 20:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2020-02-27 15:44:53.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2020-02-28 20:46:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-02-27 22:49:48.000000000","_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-02-27 15:36:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 15:44:53.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-02-27 22:49:48.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-02-28 20:46:34.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-28 21:05:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e6003c6a10a3254d210f9738552d1f6304b0005","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-02-25 15:19:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"22ba2ce4b48ab01a6f26444601143768aa2394b9","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-02-25 15:19:28.000000000","message":"Topic changed from oslo-master to oslo_drop_py2_support","accounts_in_message":[],"_revision_number":1},{"id":"019290ffb0850ea6bfcc5ab81efd843f31c166cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 15:36:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc4895832c354a9194f9107ec356fa98 : SUCCESS in 13m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e0d952d510ec4c8abfebc94506712514 : SUCCESS in 10m 21s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/b7476f3e1eff42de867116c382f0f1cc : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7513d171ef8b3d9d6055f5544f86e5a8d026ca87","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-02-25 19:42:15.000000000","message":"Patch Set 1: Workflow-1\n\nWe\u0027re going to wait this major release on removing the mox bits too.","accounts_in_message":[],"_revision_number":1},{"id":"2e311896653eacef5cde1d737966e5cd251cee9b","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-25 19:42:17.000000000","message":"Patch Set 1: Workflow-1\n\nWaiting on a final update.","accounts_in_message":[],"_revision_number":1},{"id":"f3a44d476c05766bdb9db3b49bbf585d628fe74f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-25 19:42:23.000000000","message":"Removed Workflow-1 by Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"e2b940c04b57c2b61c98dc69d99b24a82fe7b7cd","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 19:43:21.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"0073a8d2e48f90f29281adf3b070a2afb5520b2d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-27 15:34:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"065df235707af8e957e29bb8356b8c39ca82901f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-27 15:35:16.000000000","message":"Patch Set 1:\n\n\u003e We\u0027re going to wait this major release on removing the mox bits\n \u003e too.\n\ndone","accounts_in_message":[],"_revision_number":1},{"id":"955dc816bb8ee642bc51fe68a0d87be42bc9cf81","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 15:36:18.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"c188c7261665c26958616326cc00b18a43b7e400","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-27 15:44:53.000000000","message":"Patch Set 2: Code-Review+1\n\nLet\u0027s wait for Ben\u0027s appreciation","accounts_in_message":[],"_revision_number":2},{"id":"406a1b479e66c27287046eea2c24da2b6c1f194c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 16:40:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fc82ee6e65143e48ac54fdcac4a833e : SUCCESS in 13m 23s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/2e34432b8f2340b6b345368b1eaf280d : SUCCESS in 7m 16s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/ed08a89240734b1c88323c1e530755ff : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":2},{"id":"808fcc717a3947258eb9367cb3b8ab776956aa6e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2020-02-27 22:02:55.000000000","message":"Patch Set 2:\n\nWe need https://review.opendev.org/#/c/699746 first, right?","accounts_in_message":[],"_revision_number":2},{"id":"bcf4973bac7585c49bd20bbff151322d1c4f8714","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-02-27 22:49:48.000000000","message":"Patch Set 2:\n\n\u003e We need https://review.opendev.org/#/c/699746 first, right?\n\nSort of? master on that will be technically broken once we bump upper-constraints to 4.0 so we won\u0027t be able to land any other patches. I wouldn\u0027t block on it since it\u0027s trivial enough","accounts_in_message":[],"_revision_number":2},{"id":"dd33056e4ff592624ec00fd9abecb7a276507319","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-28 20:46:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n\u003e \u003e We need https://review.opendev.org/#/c/699746 first, right?\n \u003e \n \u003e Sort of? master on that will be technically broken once we bump\n \u003e upper-constraints to 4.0 so we won\u0027t be able to land any other\n \u003e patches. I wouldn\u0027t block on it since it\u0027s trivial enough\n\nAnd has now merged.","accounts_in_message":[],"_revision_number":2},{"id":"0871dff9cec0f531de883cada3087e7eec32ebac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 20:46:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c1dad1496ddf664390d2bb62f89c25235b65600d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 21:05:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bac76a382ab740dabdd2e0326ad3cf81 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d257f94e8086482b9259741ee2df9e78 : SUCCESS in 14m 29s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/158bf6d4b600460a84cde854ef61e01b : SUCCESS in 6m 46s","accounts_in_message":[],"_revision_number":2},{"id":"316195fef60a6583b4f3f5afb8b7a2f08d008805","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 21:05:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d396bc61bccb4c3d2274dbdcae66053703880e44","revisions":{"2fd2c320f0b678388f4acad32107485803cd676e":{"kind":"REWORK","_number":1,"created":"2020-02-25 15:19:08.000000000","uploader":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"ref":"refs/changes/53/709753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/709753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/709753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/709753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/709753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/709753/1"}}},"commit":{"parents":[{"commit":"2c5254e33fb679b94663759e3532cfe5339898f3","subject":"Merge \"Oslo releases for 2020-02-24\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2c5254e33fb679b94663759e3532cfe5339898f3"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-02-25 15:18:42.000000000","tz":0},"committer":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-02-25 15:18:42.000000000","tz":0},"subject":"oslotest 4.0.0","message":"oslotest 4.0.0\n\nThis release removes py2 support from oslotest.\n\nChange-Id: Ic0fbcb34a617f4dec30643b8bc5211f40bf9108f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2fd2c320f0b678388f4acad32107485803cd676e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2fd2c320f0b678388f4acad32107485803cd676e"}]},"branch":"refs/heads/master"},"d396bc61bccb4c3d2274dbdcae66053703880e44":{"kind":"REWORK","_number":2,"created":"2020-02-27 15:34:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/53/709753/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/53/709753/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/709753/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/709753/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/53/709753/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/53/709753/2"}}},"commit":{"parents":[{"commit":"819e05e3540aefa225fa32c68e442fc2e720700e","subject":"Merge \"Final release of Octavia projects for Rocky\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/819e05e3540aefa225fa32c68e442fc2e720700e"}]}],"author":{"name":"Ben Nemec","email":"bnemec@redhat.com","date":"2020-02-25 15:18:42.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2020-02-27 15:34:35.000000000","tz":0},"subject":"oslotest 4.0.0","message":"oslotest 4.0.0\n\nThis release drops py2 support and removes the \u0027oslotest.moxstubout\u0027 and\n\u0027oslotest.functional\u0027 modules.\n\nChange-Id: Ic0fbcb34a617f4dec30643b8bc5211f40bf9108f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d396bc61bccb4c3d2274dbdcae66053703880e44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d396bc61bccb4c3d2274dbdcae66053703880e44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
