)]}'
{"id":"openstack%2Freleases~709848","triplet_id":"openstack%2Freleases~master~I0779e84405efc6447dcf76aaa1f9450244d4833a","project":"openstack/releases","branch":"master","topic":"ussuri-cwi","hashtags":[],"change_id":"I0779e84405efc6447dcf76aaa1f9450244d4833a","subject":"Switch monasca-* to cycle-with-rc","status":"MERGED","created":"2020-02-25 21:59:41.000000000","updated":"2020-02-27 19:02:04.000000000","submitted":"2020-02-27 19:02:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709848-1582830124807-dc82b96e","meta_rev_id":"14ec1e76f48e3d6a8fceaa0bb428966a3144c83e","_number":709848,"virtual_id_number":709848,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-27 19:02:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-27 17:15:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-27 17:15:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-27 09:54:16.000000000","updated_by":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"reviewer":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"state":"REVIEWER"},{"updated":"2020-02-27 19:02:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7888e286142e1a9832395a1919bf69ac3c9a1aad","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-25 21:59:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ef55efbb6eb8367cf73cfc9fdbbab1aa0f249da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 22:11:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2adf8e4f996456b988f6ed1c81dcf87 : SUCCESS in 11m 18s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f2c7d8ecb03e4bd4b6db92824a8d8390 : SUCCESS in 6m 59s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/d1e398aaf1374d17a4871ba5e334d3e1 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ac55564208072a34d53a5456b1632a90abe9992b","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-02-27 09:54:16.000000000","message":"Patch Set 1:\n\n(3 comments)\n\nIn general I\u0027m OK with changing the release model. Please see notes to individual components. I can provide an update PS for these.","accounts_in_message":[],"_revision_number":1},{"id":"647be71f2c01056cc9079a48428958cc7de678fa","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 12:50:59.000000000","message":"Patch Set 1:\n\nThanks Witold. We can drop those three. If two are being retired, and one has issues with RCs, then best to avoid changing them.\n\nAnother option for the one also would be to move to independent if it is not necessarily tightly tied to the individual release cycle.","accounts_in_message":[],"_revision_number":1},{"id":"9aab734785b23ecf1eaa2f3452dfade69ed243a1","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 12:51:07.000000000","message":"Patch Set 1: Code-Review-1\n\nWill update this.","accounts_in_message":[],"_revision_number":1},{"id":"a2e70ec3b96fc29ccc319389965256a465bbf32a","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-02-27 16:33:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"fbff22e2c16a325c3806467ce80940d8d7f20075","author":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"date":"2020-02-27 16:35:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"313d56a5c6e835b7c1d884493312626359b3ab18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 17:10:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1b1be76e8d2466aaba4669e1aaade18 : SUCCESS in 15m 40s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/ad16f58d611a4e3995c9df42a00d558c : SUCCESS in 6m 28s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/f7bf91db4eb74b2c883c257b99ac7690 : SUCCESS in 7m 33s","accounts_in_message":[],"_revision_number":3},{"id":"b4e0a3dcaba0abb7c827d5c1594f6fff94a80cf4","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 17:15:29.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\nThanks!","accounts_in_message":[],"_revision_number":3},{"id":"984b7eb4ba11f3a6fcb10556829e80abd2816626","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 17:15:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"8d1661a8708e606acc18d2c6217755c28771a68e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 17:54:33.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/279ab68dc23741dfb978e43bfa366373 : POST_FAILURE in 15m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8e64445e82f45179119a943c8a3fe7c : SUCCESS in 13m 33s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e569abadcfb84ae0ad183f5a02102053 : SUCCESS in 9m 09s","accounts_in_message":[],"_revision_number":3},{"id":"94119dc8ed0b162cad86def360249d041ad007b0","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-27 17:58:34.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"ec5f19d75d2172e245a4faaea47ea6ca542bdf72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 18:21:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac71d5fa97ac4ec486798a2ad2af60b2 : SUCCESS in 13m 41s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/f675459ca6ad45c4a984fc9519eec12f : SUCCESS in 9m 06s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/6ca13f5102264abb9418da8e336b179e : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":3},{"id":"c8b51de1e6c231870d9ed5d6e47735d462f04d9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 18:21:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"a5f3c0ed836800a987a0bff811b0a97511071e67","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 19:02:04.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/299a4fcedb7146e494e9142d563f5d35 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1310072959e84eaa93d0352846374d39 : SUCCESS in 12m 27s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/6be4ca0f7ae641d683adb73a7b5e8e81 : SUCCESS in 8m 37s","accounts_in_message":[],"_revision_number":3},{"id":"174c8ce88d5d5033cab4c1503ed40fbbd1f1ff81","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-27 19:02:04.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"25a08d003e22f6306b00420cfd80317ca352b9f8","revisions":{"37656d8e3efbb6bdfec8350e91dc04ffa71e0ac6":{"kind":"REWORK","_number":1,"created":"2020-02-25 21:59:41.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/48/709848/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/48/709848/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/709848/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/709848/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/709848/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/48/709848/1"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-25 21:59:26.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-25 21:59:26.000000000","tz":-360},"subject":"Switch monasca-* to cycle-with-rc","message":"Switch monasca-* to cycle-with-rc\n\nDeliverables following the cycle-with-intermediary model are expected\nto release multiple times throughout the development cycle. Since\nmonasca-* only released once last cycle, and has not done a release\nyet this cycle, this proposes changing the release model to cycle-with-rc\nso the expectation is only that it will release an RC by the end of\nthe cycle prior to doing the final release.\n\nFor the team, please respond to this patch in one of the following ways:\n\n* -1 this patch and follow up with an intermediary release\n* +1 this patch and we will switch over the release model\n* -1 this patch, but acknowledge a release will be done prior to the\n  RC1 deadline.\n\nChange-Id: I0779e84405efc6447dcf76aaa1f9450244d4833a\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/37656d8e3efbb6bdfec8350e91dc04ffa71e0ac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/37656d8e3efbb6bdfec8350e91dc04ffa71e0ac6"}]},"branch":"refs/heads/master"},"15b37748cbda8b7104030ec2aad9e47ab07c5d16":{"kind":"NO_CHANGE","_number":2,"created":"2020-02-27 16:33:29.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/48/709848/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/48/709848/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/709848/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/709848/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/709848/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/48/709848/2"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-25 21:59:26.000000000","tz":-360},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-27 16:33:02.000000000","tz":60},"subject":"Switch monasca-* to cycle-with-rc","message":"Switch monasca-* to cycle-with-rc\n\nDeliverables following the cycle-with-intermediary model are expected\nto release multiple times throughout the development cycle. Since\nmonasca-* only released once last cycle, and has not done a release\nyet this cycle, this proposes changing the release model to cycle-with-rc\nso the expectation is only that it will release an RC by the end of\nthe cycle prior to doing the final release.\n\nFor the team, please respond to this patch in one of the following ways:\n\n* -1 this patch and follow up with an intermediary release\n* +1 this patch and we will switch over the release model\n* -1 this patch, but acknowledge a release will be done prior to the\n  RC1 deadline.\n\nChange-Id: I0779e84405efc6447dcf76aaa1f9450244d4833a\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/15b37748cbda8b7104030ec2aad9e47ab07c5d16"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/15b37748cbda8b7104030ec2aad9e47ab07c5d16"}]},"branch":"refs/heads/master"},"25a08d003e22f6306b00420cfd80317ca352b9f8":{"kind":"REWORK","_number":3,"created":"2020-02-27 16:35:56.000000000","uploader":{"_account_id":16222,"name":"witek","email":"witold.bedyk@suse.com","username":"witek"},"ref":"refs/changes/48/709848/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/48/709848/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/709848/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/709848/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/48/709848/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/48/709848/3"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-25 21:59:26.000000000","tz":-360},"committer":{"name":"Witek Bedyk","email":"witold.bedyk@suse.com","date":"2020-02-27 16:35:34.000000000","tz":60},"subject":"Switch monasca-* to cycle-with-rc","message":"Switch monasca-* to cycle-with-rc\n\nDeliverables following the cycle-with-intermediary model are expected\nto release multiple times throughout the development cycle. Since\nmonasca-* only released once last cycle, and has not done a release\nyet this cycle, this proposes changing the release model to cycle-with-rc\nso the expectation is only that it will release an RC by the end of\nthe cycle prior to doing the final release.\n\nFor the team, please respond to this patch in one of the following ways:\n\n* -1 this patch and follow up with an intermediary release\n* +1 this patch and we will switch over the release model\n* -1 this patch, but acknowledge a release will be done prior to the\n  RC1 deadline.\n\nChange-Id: I0779e84405efc6447dcf76aaa1f9450244d4833a\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25a08d003e22f6306b00420cfd80317ca352b9f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/25a08d003e22f6306b00420cfd80317ca352b9f8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
