)]}'
{"id":"openstack%2Freleases~709852","triplet_id":"openstack%2Freleases~master~I3a3fdf8ed1fccff4090c7e3c144ade52ab8d1167","project":"openstack/releases","branch":"master","topic":"ussuri-cwi","hashtags":[],"change_id":"I3a3fdf8ed1fccff4090c7e3c144ade52ab8d1167","subject":"Switch senlin-dashboard to cycle-with-rc","status":"MERGED","created":"2020-02-25 22:05:29.000000000","updated":"2020-03-12 14:04:13.000000000","submitted":"2020-03-12 14:04:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709852-1584021853821-962e1dba","meta_rev_id":"6d33f8224f3c98cb6f819aed5de1d86372f902b1","_number":709852,"virtual_id_number":709852,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"date":"2020-03-12 10:10:37.000000000","_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-12 14:04:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-12 13:52:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":1,"date":"2020-03-11 00:55:06.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},{"value":1,"date":"2020-03-12 10:21:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":1,"date":"2020-02-28 11:42:42.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-12 13:52:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"date":"2020-02-25 22:28:05.000000000","_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},{"value":0,"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"value":0,"date":"2020-03-12 10:10:18.000000000","_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-03-12 10:23:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 22:28:05.000000000","updated_by":{"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},"reviewer":{"_account_id":8246,"name":"Qiming Teng","email":"tengqm@outlook.com","username":"tengqm"},"state":"REVIEWER"},{"updated":"2020-02-28 11:42:42.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-11 00:55:06.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2020-03-12 10:10:18.000000000","updated_by":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"reviewer":{"_account_id":27224,"name":"Duc Truong","email":"duc.openstack@gmail.com","username":"dtruong2"},"state":"REVIEWER"},{"updated":"2020-03-12 10:10:37.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-03-12 10:21:42.000000000","updated_by":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"reviewer":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"state":"REVIEWER"},{"updated":"2020-03-12 14:04:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8faa414937663b71c4a4ccb1a54607737bb2cc0d","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-25 22:05:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f37e99c3dd159128f2bfd2212eb5d60a55f37efb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 22:19:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60963df948524b9d9f2d059267f9e5bd : SUCCESS in 12m 32s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/696bdf6780294b42b1a9c9421dcd3f5e : SUCCESS in 6m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/334f0f5b607f4b66bc3857f83b8263b6 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"da8f8c8de5bcaa5d723aeef956d5f66c15f315bd","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-28 11:42:42.000000000","message":"Patch Set 1: Code-Review+1\n\nThe rationale behind these changes make sense.\n\nLGTM only +1 for now by waiting for PLT/liaison/team approval.","accounts_in_message":[],"_revision_number":1},{"id":"756d1ecac2ac54a374388a684458ae1287934ada","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-03-11 00:55:06.000000000","message":"Patch Set 1: Code-Review+1\n\nWaiting for PTL/liaison ack.","accounts_in_message":[],"_revision_number":1},{"id":"3e02a1b511c7906b99c572e894a6e3c7b365844a","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-03-12 10:10:37.000000000","message":"Patch Set 1:\n\nAdding PTL since he was missing in CC:","accounts_in_message":[],"_revision_number":1},{"id":"d0adc0ad6c13f37d3ead46c80b6c70fb0243f746","author":{"_account_id":22998,"name":"XueFengLiu","email":"liu.xuefeng1@zte.com.cn","username":"sgfeng"},"date":"2020-03-12 10:21:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"059b1201e125c00dd4d294fa54bb047e7226826b","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 10:23:24.000000000","message":"Patch Set 1: PTL-Approved+1","accounts_in_message":[],"_revision_number":1},{"id":"f462953bbd552b39ea8a9ab0d10a915926068f8f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-12 13:52:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fddbc658b60ac12ded7a193c0ad44a0c48fc1dbe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 13:53:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e98548331b3bc3e16546108151add4e27ec5072c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 14:04:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9626f6aee0f7488d2dc1774d12d26ff7d7cc3944","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-12 14:04:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e338533ccf4f460e9bf96f4120001665 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897ac79aa7cc47488c715f2fd6838088 : SUCCESS in 10m 13s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e8affb6c7f9d493cada0f084b0dfce7b : SUCCESS in 6m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c37fe8f690cf8f27ea52f0ad96fdd62d381b7a1","revisions":{"0c37fe8f690cf8f27ea52f0ad96fdd62d381b7a1":{"kind":"REWORK","_number":1,"created":"2020-02-25 22:05:29.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/52/709852/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/52/709852/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/709852/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/709852/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/52/709852/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/52/709852/1"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-25 22:05:25.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-25 22:05:25.000000000","tz":-360},"subject":"Switch senlin-dashboard to cycle-with-rc","message":"Switch senlin-dashboard to cycle-with-rc\n\nDeliverables following the cycle-with-intermediary model are expected\nto release multiple times throughout the development cycle. Since\nsenlin-dashboard only released once last cycle, and has not done a release\nyet this cycle, this proposes changing the release model to cycle-with-rc\nso the expectation is only that it will release an RC by the end of\nthe cycle prior to doing the final release.\n\nFor the team, please respond to this patch in one of the following ways:\n\n* -1 this patch and follow up with an intermediary release\n* +1 this patch and we will switch over the release model\n* -1 this patch, but acknowledge a release will be done prior to the\n  RC1 deadline.\n\nChange-Id: I3a3fdf8ed1fccff4090c7e3c144ade52ab8d1167\nSigned-off-by: Sean McGinnis \u003csean.mcginnis@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0c37fe8f690cf8f27ea52f0ad96fdd62d381b7a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/0c37fe8f690cf8f27ea52f0ad96fdd62d381b7a1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
