)]}'
{"id":"openstack%2Freleases~709875","triplet_id":"openstack%2Freleases~master~I121b241b4457459dbfe4edfd42af08162fa076e2","project":"openstack/releases","branch":"master","topic":"rocky-em","hashtags":[],"change_id":"I121b241b4457459dbfe4edfd42af08162fa076e2","subject":"[OpenStackAnsible] Transition Rocky to EM","status":"MERGED","created":"2020-02-26 00:59:16.000000000","updated":"2020-03-31 14:28:24.000000000","submitted":"2020-03-31 14:28:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709875-1585664904459-6894be15","meta_rev_id":"8997bc473ed22cec3d23e7fb61664cbfb881673a","_number":709875,"virtual_id_number":709875,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-31 14:28:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-30 14:41:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2020-03-30 15:06:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-31 14:12:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 00:59:16.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2020-03-03 16:32:21.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-03-03 16:55:57.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"},{"updated":"2020-03-19 16:36:51.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-03-19 16:40:44.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-03-30 15:06:26.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-31 14:28:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48ad4245034b20d6cce303b9302baa7522079994","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 00:59:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8da2ae8d637ac3781d55efa22ee16d4ee5fcf4e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 01:11:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff55276cf8d4095ad5a339e519c9f94 : SUCCESS in 11m 44s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/e4c36864a31942d59044a6d7a85bc9cd : SUCCESS in 6m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/9c113d2b658d4fcab846a4780f8721df : SUCCESS in 7m 38s","accounts_in_message":[],"_revision_number":1},{"id":"80c579e52eed5f313a3437acba1962f2fbef85ec","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-26 17:45:39.000000000","message":"Patch Set 1: Code-Review+1\n\nspecial beast. ok if the team is okay.","accounts_in_message":[],"_revision_number":1},{"id":"26261486cf439f6e5933abb3b734751013e12f7e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-02-27 13:40:02.000000000","message":"Patch Set 1: Code-Review-1\n\nYep, I\u0027d love to place the final release tbh. We need to merge https://review.opendev.org/#/c/708083/ and https://review.opendev.org/#/c/706709/ for that...","accounts_in_message":[],"_revision_number":1},{"id":"10fed53201cb20001349f91a6aca59338896b2b8","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-28 08:23:33.000000000","message":"Removed Code-Review+1 by Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","accounts_in_message":[],"_revision_number":1},{"id":"a68e929e6e112319e26070d120d6594a2e3ef4ad","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-02-28 20:02:16.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"bd6106b9eb87e3f30db09e9be5866b31baf4584c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-02-28 20:02:21.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"228f33eb6ae4a61d748b1430b2857dc4751a6e01","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 20:04:05.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"b6f3dc986b8e5134b90ac6cbe519900a32414f63","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-28 20:04:52.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"30036e6b328f0c339a15154fedcd51804a0412b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 20:19:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/619e53adcaa340d49bb6a25af2d1007e : SUCCESS in 12m 43s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/11ee9bc24c604872a5ef71c249ead692 : SUCCESS in 7m 44s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/e1f2ca2495a14fbba71b0d514fbe3869 : SUCCESS in 7m 43s","accounts_in_message":[],"_revision_number":2},{"id":"49c064db9b3d084d22c630017aecd2b23871ec5a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-02-28 20:19:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a2e08facf02389e0b1f0dd711652075cbc5d3569","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-28 20:29:02.000000000","message":"Patch Set 3:\n\nI think you had a couple things you wanted to get released first, right? You will need to propose a release for those. Then once that is done, this can be rebased on top of that and the final rocky-em hash can be updated to match whatever that final release was.","accounts_in_message":[],"_revision_number":3},{"id":"552e1779a14fb339892455a98fdd92a1bfc12bfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-28 20:39:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c660a211a57487eb44692fcf3056e83 : SUCCESS in 13m 57s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1c4657701003498c899dcc782943bcf2 : FAILURE in 12m 19s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/87fd485985a24ed58d5ad840c72b1d8d : SUCCESS in 6m 53s","accounts_in_message":[],"_revision_number":3},{"id":"cfd3511a71525ea69d893dcba05052f9d38aab25","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-02 14:08:59.000000000","message":"Patch Set 3:\n\nI think I\u0027ve already rebased, so waiting for the release to be merged to recheck this patch","accounts_in_message":[],"_revision_number":3},{"id":"9e60b3592f64a7706db856966d98e2ea3e3ccf11","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-02 15:50:50.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"915f5e8492b97a88c2323076a7661273b1dbfbf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 16:07:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/781490d2bdcb4da7b0d4ce45efa2eb86 : SUCCESS in 14m 51s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/1edf6d03a7714ff89fa0a9a0c6b32956 : SUCCESS in 8m 54s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/784224e6e53544ca9e3721e2f053befd : SUCCESS in 9m 56s","accounts_in_message":[],"_revision_number":3},{"id":"3aa0612834fae279fb9d45a6bde35df093f6f5fe","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-02 17:24:32.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"282b34ca4e8530dc0d535dac0b6b28b9ab2b19fc","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-02 17:25:59.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"c2f90679c7990866428ee7f623c763b44efbf2fb","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-03-03 16:32:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d947abcf87fe97a04acd04d4f3a401b29fdbfcc2","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2020-03-03 16:55:57.000000000","message":"Patch Set 3:\n\nI think these need to get in by the last release;\n\nhttps://review.opendev.org/#/c/708115/\nhttps://review.opendev.org/#/c/708420/\n\nIt would ease the transition from xenial to bionic. I also wrote a doc on this transition.\n\nhttps://review.opendev.org/#/c/710803/","accounts_in_message":[],"_revision_number":3},{"id":"f1c7717416b71e28bacf1a810af580e6ebb90476","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-04 10:13:35.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"706bb697325b8d7ab2f86ec5718a19dd68f81cdf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-19 16:36:51.000000000","message":"Patch Set 3: Code-Review-1\n\nok, let\u0027s include these patches to EM","accounts_in_message":[],"_revision_number":3},{"id":"1913cc82f13c2ea558b915c6c6af9ba2dc01cf4d","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 16:39:17.000000000","message":"Patch Set 3: -PTL-Approved","accounts_in_message":[],"_revision_number":3},{"id":"a46dc45c65dfcd00214c435ba1be04679bca5ea2","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-03-19 16:40:44.000000000","message":"Patch Set 3:\n\n\u003e ok, let\u0027s include these patches to EM\n\nThanks for updating us! And thanks for the work you\u0027re doing for the community, improving docs and release notes are very good before an EM.","accounts_in_message":[],"_revision_number":3},{"id":"62d7114f373d79db0872a2409182e6eaca89d607","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-24 21:23:30.000000000","message":"Patch Set 3:\n\n\u003e I think these need to get in by the last release;\n \u003e \n \u003e https://review.opendev.org/#/c/708115/\n \u003e https://review.opendev.org/#/c/708420/\n \u003e \n \u003e It would ease the transition from xenial to bionic. I also wrote a\n \u003e doc on this transition.\n \u003e \n \u003e https://review.opendev.org/#/c/710803/\n\nLooks like the first two have made it in. The last one should be easy to get in quick too, but doesn\u0027t look like there\u0027s been any activity on it.\n\nWe should get a final release out and get this wrapped up. We are a few weeks past the published transition date now.","accounts_in_message":[],"_revision_number":3},{"id":"dae610c04501b29389b6fbfbcdc4dde2fb119977","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-25 11:06:48.000000000","message":"Patch Set 3: Code-Review+1\n\nWaiting for validation","accounts_in_message":[],"_revision_number":3},{"id":"b48d41767bd8fe90af87a4201f2daee82e5792c3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-30 14:38:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f3e034fb0eb7e2309f9ab8466558c590a4e8297f","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-30 14:41:55.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)\n\nThanks!","accounts_in_message":[],"_revision_number":4},{"id":"d7c0c4e2196a0a86f39f01c684a59119e635e2bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-30 14:57:09.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65a399c35b8a43389354a16fcc2f1fc6 : SUCCESS in 12m 55s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d1a4c6bcf0e74633beeb24b34c73fab5 : SUCCESS in 7m 29s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/3a607e75b71e41b6885aa7a790c45d4b : SUCCESS in 8m 42s","accounts_in_message":[],"_revision_number":4},{"id":"ba65934c94c86f511889e230ba8e56d67fdbad4b","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-30 15:06:26.000000000","message":"Patch Set 4: Code-Review+2\n\nWaiting for PTL/team approval","accounts_in_message":[],"_revision_number":4},{"id":"c63b6537649cb1f14516ba6a52ea5424589bcf68","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-31 14:12:14.000000000","message":"Patch Set 4: Workflow+1\n\nShould be good with the last release. Good to wrap up rocky EM transition.","accounts_in_message":[],"_revision_number":4},{"id":"206f1cbea01aa46852953e44cc7a7f30c8710981","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:12:27.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0268ba1932f1dc31127c6acac903a262cb443114","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:28:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"a4abb3af568e461cc5522e10abd71e5d896ee64b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-31 14:28:24.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3639182a9846b095e0f8cfc3bb15bd : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1baac1e87a264744b9b45d9ed860efd7 : SUCCESS in 10m 14s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/a9bb99992f684f5c93c33f25e9d7d44c : SUCCESS in 6m 57s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2d24ffc4201f3e09297d40599f683ea86e8e5487","revisions":{"ca85aadbbb80daeeacf30944a02d9e4268f8e80c":{"kind":"REWORK","_number":1,"created":"2020-02-26 00:59:16.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/75/709875/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/709875/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/709875/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/709875/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/709875/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/709875/1"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:46:14.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:46:14.000000000","tz":-360},"subject":"[OpenStackAnsible] Transition Rocky to EM","message":"[OpenStackAnsible] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I121b241b4457459dbfe4edfd42af08162fa076e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ca85aadbbb80daeeacf30944a02d9e4268f8e80c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/ca85aadbbb80daeeacf30944a02d9e4268f8e80c"}]},"branch":"refs/heads/master"},"a917d782b2abb2eec9f49d255e8c9d8d84d2c7af":{"kind":"REWORK","_number":2,"created":"2020-02-28 20:02:16.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/75/709875/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/709875/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/709875/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/709875/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/709875/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/709875/2"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:46:14.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-02-28 19:07:03.000000000","tz":0},"subject":"[OpenStackAnsible] Transition Rocky to EM","message":"[OpenStackAnsible] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I121b241b4457459dbfe4edfd42af08162fa076e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a917d782b2abb2eec9f49d255e8c9d8d84d2c7af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/a917d782b2abb2eec9f49d255e8c9d8d84d2c7af"}]},"branch":"refs/heads/master"},"e653113462a83503453e3f255478368c8fec9759":{"kind":"REWORK","_number":3,"created":"2020-02-28 20:19:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/75/709875/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/709875/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/709875/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/709875/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/709875/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/709875/3"}}},"commit":{"parents":[{"commit":"235b3291021beebf4bbda3e8008a3b1a61deb24d","subject":"Release OpenStack-Ansible Rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/235b3291021beebf4bbda3e8008a3b1a61deb24d"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:46:14.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-02-28 20:19:37.000000000","tz":120},"subject":"[OpenStackAnsible] Transition Rocky to EM","message":"[OpenStackAnsible] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I121b241b4457459dbfe4edfd42af08162fa076e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e653113462a83503453e3f255478368c8fec9759"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/e653113462a83503453e3f255478368c8fec9759"}]},"branch":"refs/heads/master"},"2d24ffc4201f3e09297d40599f683ea86e8e5487":{"kind":"REWORK","_number":4,"created":"2020-03-30 14:38:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/75/709875/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/75/709875/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/709875/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/709875/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/75/709875/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/75/709875/4"}}},"commit":{"parents":[{"commit":"7b4aac601cdd3c8d147fd9e029a42d8a756c89ce","subject":"Merge \"Release OSA Rocky\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7b4aac601cdd3c8d147fd9e029a42d8a756c89ce"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:46:14.000000000","tz":-360},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-03-30 14:38:11.000000000","tz":180},"subject":"[OpenStackAnsible] Transition Rocky to EM","message":"[OpenStackAnsible] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I121b241b4457459dbfe4edfd42af08162fa076e2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d24ffc4201f3e09297d40599f683ea86e8e5487"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/2d24ffc4201f3e09297d40599f683ea86e8e5487"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
