)]}'
{"id":"openstack%2Freleases~709888","triplet_id":"openstack%2Freleases~master~I597132f63465e888a22889e4d7663a33f29d21aa","project":"openstack/releases","branch":"master","topic":"rocky-em","hashtags":[],"change_id":"I597132f63465e888a22889e4d7663a33f29d21aa","subject":"[glance] Transition Rocky to EM","status":"MERGED","created":"2020-02-26 01:00:07.000000000","updated":"2020-03-19 21:39:26.000000000","submitted":"2020-03-19 21:39:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709888-1584653966819-baf8bc82","meta_rev_id":"5a7824712df24ce1f91fd3ad6d98b894a575ffce","_number":709888,"virtual_id_number":709888,"owner":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-19 21:39:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-03-19 21:24:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2020-03-19 16:05:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":1,"date":"2020-03-19 16:15:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-03-19 21:24:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"PTL-Approved":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:release-approval","value":1,"date":"2020-03-19 16:24:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{" 0":"Not approved yet","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-26 01:00:07.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2020-02-27 10:45:46.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2020-03-04 12:19:20.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2020-03-17 17:20:35.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2020-03-19 16:05:52.000000000","updated_by":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"reviewer":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"state":"REVIEWER"},{"updated":"2020-03-19 16:15:43.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2020-03-19 21:39:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a171653439f0b3887fdaa3815781a29472f700ef","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 01:00:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ee39f59c395f11ac6c08d8b3b6f44fc05f767d9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 01:13:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13b35ff8acf848cc913db87c4fac9480 : SUCCESS in 12m 37s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/5b1dea61703b4d7284488c695d8f02a1 : SUCCESS in 6m 20s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/64af039a73854949b1d141a9bca21b1d : SUCCESS in 6m 49s","accounts_in_message":[],"_revision_number":1},{"id":"25c1017763c47e9baa2329e6bec265a43eb470b9","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-26 17:40:09.000000000","message":"Patch Set 1: Code-Review-1\n\npython-glanceclient missing.","accounts_in_message":[],"_revision_number":1},{"id":"a97cc62bb438709e40bbd48081d3c19b036c8987","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-26 17:42:23.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e02596031a491352e83f4d056f3567c2abb92600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-26 17:55:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/835b2550eb284339b720ebf5eeb24525 : SUCCESS in 12m 00s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d6cb9024232a4908b18322f08ffb98ef : SUCCESS in 7m 03s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/56048f2ad9f3427098e3ae2829376897 : SUCCESS in 6m 47s","accounts_in_message":[],"_revision_number":2},{"id":"9d19bfc3177c48bc8ab77351c1a571b3d3b40da4","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2020-02-27 10:45:46.000000000","message":"Patch Set 2:\n\nSee unreleased but merged changes: https://etherpad.openstack.org/p/glance-stable-rocky-em","accounts_in_message":[],"_revision_number":2},{"id":"0a12218b7594ad08135ac6a8b2c86dd0cd50df36","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-02-27 14:27:01.000000000","message":"Patch Set 2: Code-Review+1\n\nLGTM in the current state, waiting for updates/validations.","accounts_in_message":[],"_revision_number":2},{"id":"7154779ae4c9cc7ea48ffcd86f513999398d9b95","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-02-28 08:30:11.000000000","message":"Patch Set 2:\n\n\u003e See unreleased but merged changes: https://etherpad.openstack.org/p/glance-stable-rocky-em\n\nThanks Elod for your work!","accounts_in_message":[],"_revision_number":2},{"id":"5695e413ca550a464ebb715757ebf879ef9164f2","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2020-03-04 12:19:20.000000000","message":"Patch Set 2:\n\nThere are lots of things that won\u0027t be included, should we really do this?\n\nSee also https://zuul.opendev.org/t/openstack/build/56048f2ad9f3427098e3ae2829376897/log/tox/list-changes-results.log#680","accounts_in_message":[],"_revision_number":2},{"id":"48a500adfda876a08511ba14c4ce4f25b8bbf8f2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-11 05:01:35.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"cadbb2f452f0bdb0085717a279bf0228c3f5e3d3","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-11 05:02:40.000000000","message":"Patch Set 2: PTL-Approved+1","accounts_in_message":[],"_revision_number":2},{"id":"a2fcc7aa0f411ad8d6561a35edee5e95c438ffe6","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-11 10:02:54.000000000","message":"Patch Set 2:\n\n\u003e There are lots of things that won\u0027t be included, should we really\n \u003e do this?\n \u003e \n \u003e See also https://zuul.opendev.org/t/openstack/build/56048f2ad9f3427098e3ae2829376897/log/tox/list-changes-results.log#680\n\nAny updates about this ^^^^^^?","accounts_in_message":[],"_revision_number":2},{"id":"6f7835d111158b855723e19c2e90080f2244468b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-12 04:33:24.000000000","message":"Patch Set 2:\n\n\u003e \u003e There are lots of things that won\u0027t be included, should we really\n \u003e \u003e do this?\n \u003e \u003e\n \u003e \u003e See also https://zuul.opendev.org/t/openstack/build/56048f2ad9f3427098e3ae2829376897/log/tox/list-changes-results.log#680\n \u003e \n \u003e Any updates about this ^^^^^^?\n\nWill discuss this in today\u0027s meeting and update here.","accounts_in_message":[],"_revision_number":2},{"id":"1c72247046b32fd4f7fc5caf47344bed8bab8a48","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-17 12:59:40.000000000","message":"Patch Set 2: Code-Review+2\n\n\u003e \u003e \u003e There are lots of things that won\u0027t be included, should we\n \u003e really\n \u003e \u003e \u003e do this?\n \u003e \u003e \u003e\n \u003e \u003e \u003e See also https://zuul.opendev.org/t/openstack/build/56048f2ad9f3427098e3ae2829376897/log/tox/list-changes-results.log#680\n \u003e \u003e\n \u003e \u003e Any updates about this ^^^^^^?\n \u003e \n \u003e Will discuss this in today\u0027s meeting and update here.\n\nAny update?\n\nI suppose it\u0027s good for you since you already +1\u0027ed, so I move to +2 but I leave the +W to 0","accounts_in_message":[],"_revision_number":2},{"id":"af44ff505688210dd9c34b3e53d677570948e774","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-17 17:11:41.000000000","message":"Patch Set 2: -Code-Review\n\n(3 comments)\n\nWe have discussed this in last weeks glance weekly meeting and decided to -2 current open patches which are open and not priority for us. I have updated the etherpad https://etherpad.openstack.org/p/glance-stable-rocky-em with current status of mentioned patches.\n\nPlease compare the hash, I have added latest one in the comments.","accounts_in_message":[],"_revision_number":2},{"id":"2038c1e490a1df227af65e8ab11ff6344d6281fe","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-17 17:12:58.000000000","message":"Patch Set 2: -PTL-Approved","accounts_in_message":[],"_revision_number":2},{"id":"6d95dfa426599815c7591f4b971dc5e7161c13bc","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2020-03-17 17:20:35.000000000","message":"Patch Set 2: -Code-Review\n\n\u003e (3 comments)\n \u003e \n \u003e We have discussed this in last weeks glance weekly meeting and\n \u003e decided to -2 current open patches which are open and not priority\n \u003e for us. I have updated the etherpad https://etherpad.openstack.org/p/glance-stable-rocky-em\n \u003e with current status of mentioned patches.\n \u003e \n \u003e Please compare the hash, I have added latest one in the comments.\n\nThanks for your feedback","accounts_in_message":[],"_revision_number":2},{"id":"1b8d46fba0b1d039faaf812520db9633cd08ab65","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-17 18:08:31.000000000","message":"Patch Set 2:\n\nLooks like most have merged. Do you plan on getting a final release done soon Abhishek?","accounts_in_message":[],"_revision_number":2},{"id":"378f2fb64b3a7400e42e6eb1800a6a0d2fc009bc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-19 05:37:00.000000000","message":"Patch Set 2:\n\n\u003e Looks like most have merged. Do you plan on getting a final release\n \u003e done soon Abhishek?\nHi Sean, \n\nglance - https://review.opendev.org/713779\nglance_store - https://review.opendev.org/713782\npython-glanceclient - https://review.opendev.org/713783","accounts_in_message":[],"_revision_number":2},{"id":"8ab6ddc5c2e03f299fc2f6da9f83a9ba194fbe02","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-19 15:22:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6cd89e638271d5684d1511c869bde427db362512","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 15:37:10.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9931ad7407f54429aab5bb9baf8c997d : SUCCESS in 12m 52s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/3abea8307bfb4421927319f223d155f8 : SUCCESS in 9m 40s\n- releases-tox-list-changes https://zuul.opendev.org/t/openstack/build/c49b5776fbc94c96917436488e44ff51 : SUCCESS in 6m 06s","accounts_in_message":[],"_revision_number":3},{"id":"7a7fa433812f4a3d3844f600935196cbf1b1afc0","author":{"_account_id":308,"name":"Thierry Carrez","email":"thierry@openstack.org","username":"ttx"},"date":"2020-03-19 16:05:52.000000000","message":"Patch Set 3: Code-Review+2\n\nWould be great to have Abishek\u0027s +1","accounts_in_message":[],"_revision_number":3},{"id":"a153de2407eb89ddc60a669967654958ede9df26","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2020-03-19 16:15:43.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bfce69560f1b3302d0505acfe1bafe722e5a2de2","tag":"autogenerated:zuul:release-approval","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 16:24:35.000000000","message":"Patch Set 3: PTL-Approved+1","accounts_in_message":[],"_revision_number":3},{"id":"787379e9e600316a8b41c37f5ff813aae6ebd859","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-03-19 21:24:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3cc7f117a99398d9593cbaf13fe4e60885957ca9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 21:24:27.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b2f8c97e75170fcb4c4a862ede4d1ffb5e6a8a5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 21:39:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f775ebe3bd554d4194c807be593a3baf : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6fc4659df4c4d1ab19c374c6ad45210 : SUCCESS in 10m 06s\n- openstack-tox-validate https://zuul.opendev.org/t/openstack/build/d37832a0c68b496ca57fcca492d3f8dd : SUCCESS in 6m 39s","accounts_in_message":[],"_revision_number":3},{"id":"4d017e11a4829fba77796aaf48a122335b572e5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-19 21:39:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d433a8df8525f3a871104ed676dda98cabd8e182","revisions":{"14024a50470e8f5d9b02cccb645b777afd32d5b4":{"kind":"REWORK","_number":1,"created":"2020-02-26 01:00:07.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/88/709888/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/709888/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/709888/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/709888/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/709888/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/709888/1"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:42:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:42:45.000000000","tz":-360},"subject":"[glance] Transition Rocky to EM","message":"[glance] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I597132f63465e888a22889e4d7663a33f29d21aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14024a50470e8f5d9b02cccb645b777afd32d5b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/14024a50470e8f5d9b02cccb645b777afd32d5b4"}]},"branch":"refs/heads/master"},"7e0bdc2702b1525890471f13a0b125c98487d39f":{"kind":"REWORK","_number":2,"created":"2020-02-26 17:42:23.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/88/709888/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/709888/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/709888/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/709888/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/709888/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/709888/2"}}},"commit":{"parents":[{"commit":"16f54d6854dd9291772dfc9f8b508df926cda551","subject":"Merge \"neutron-lbaas-dashboard is a deprecated deliverable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/16f54d6854dd9291772dfc9f8b508df926cda551"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:42:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 17:42:19.000000000","tz":-360},"subject":"[glance] Transition Rocky to EM","message":"[glance] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I597132f63465e888a22889e4d7663a33f29d21aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e0bdc2702b1525890471f13a0b125c98487d39f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/7e0bdc2702b1525890471f13a0b125c98487d39f"}]},"branch":"refs/heads/master"},"d433a8df8525f3a871104ed676dda98cabd8e182":{"kind":"REWORK","_number":3,"created":"2020-03-19 15:22:21.000000000","uploader":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"ref":"refs/changes/88/709888/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/releases","ref":"refs/changes/88/709888/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/709888/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/709888/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/releases refs/changes/88/709888/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/releases refs/changes/88/709888/3"}}},"commit":{"parents":[{"commit":"3b1786928e8325e9b36c170bec68fb6ff9020537","subject":"Merge \"Release python-glanceclient 2.13.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/3b1786928e8325e9b36c170bec68fb6ff9020537"}]}],"author":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-02-26 00:42:45.000000000","tz":-360},"committer":{"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","date":"2020-03-19 15:22:12.000000000","tz":-300},"subject":"[glance] Transition Rocky to EM","message":"[glance] Transition Rocky to EM\n\nThis transition the rocky branch to extended maintenance.\nChanges for bugfixes and things the team deems important are\nstill encouraged, but there will no longer be official releases\noff of the branch.\n\nPlease +1 if the team is ready for us to proceed with this\ntransition, or -1 if there are any final backports currently in\nflight that we should wait for. For the latter case, please\nupdate the patch with the new commit hash after doing a final\nrelease to get those changes out so we know to proceed with the\ntransition.\n\nChange-Id: I597132f63465e888a22889e4d7663a33f29d21aa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d433a8df8525f3a871104ed676dda98cabd8e182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/releases/commit/d433a8df8525f3a871104ed676dda98cabd8e182"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
